)]}'
{"id":"openstack%2Fheat-tempest-plugin~823794","triplet_id":"openstack%2Fheat-tempest-plugin~master~I66de53e9f02d53b5c36f1b880c00106372ddb582","project":"openstack/heat-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-09-16 09:13:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I66de53e9f02d53b5c36f1b880c00106372ddb582","subject":"DNM testing out the plugin","status":"ABANDONED","created":"2022-01-07 12:51:02.000000000","updated":"2024-09-16 09:13:39.000000000","total_comment_count":3,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"01775ac4d3b29a406b35cdf7ecf18ab7807e0695","_number":823794,"virtual_id_number":823794,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-07 14:01:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-01-07 14:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 23:46:17.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"149fc2fa7677b7f8c3f7ff72d343c4962b75c0f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-07 12:51:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e899dd8d22f504b339aa9e277b2e33f4c187b921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:01:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3486a38181342e9b4ed764200ad132a : SUCCESS in 3m 19s\n- heat-functional https://zuul.opendev.org/t/openstack/build/73c7292a0498449e96894e20b46dad6e : SUCCESS in 53m 48s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/3d2387aaac42408a93881fb2030eaa49 : SUCCESS in 47m 33s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/2f31367ae60840f1ad50d754ad1ed05b : FAILURE in 32m 34s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/5ef138316a9f494f9b3048298ff611c2 : FAILURE in 37m 16s (non-voting)\n- heat-functional-ussuri https://zuul.opendev.org/t/openstack/build/35ac561697714124b6bcafd96e7fd962 : SUCCESS in 55m 30s\n- heat-functional-victoria https://zuul.opendev.org/t/openstack/build/d4095d230b65495e860133f2740422ef : SUCCESS in 1h 07m 10s\n- heat-functional-wallaby https://zuul.opendev.org/t/openstack/build/5e96f6651ec24563a18935f3b9d27c90 : SUCCESS in 59m 36s\n- heat-functional-xena https://zuul.opendev.org/t/openstack/build/56bd4e21289f48ab91d413a698b28f36 : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7ffbcf849e1d990857951890db68f3e8c9a647eb","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-17 14:32:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4979531dc5639c8a6b16ffcc2a03cb56783f6ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-17 15:40:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba5e0690e194b86bd66e3066a1660a5 : SUCCESS in 4m 20s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d426d452e491488790ec2fa1e2924978 : FAILURE in 57m 40s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d123bbf3db76404bab31becdcc7e63e2 : FAILURE in 46m 10s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/e6055531695349b2b2c72da1e8abc44d : FAILURE in 34m 19s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/6de6b5defa3341b3b571966bc1c9d195 : FAILURE in 47m 43s (non-voting)\n- heat-functional-ussuri https://zuul.opendev.org/t/openstack/build/b3c1bfb9a147463aab697ede229ea974 : FAILURE in 1h 01m 04s\n- heat-functional-victoria https://zuul.opendev.org/t/openstack/build/dcaab489a06646cf91bfb794e00b0ece : FAILURE in 59m 06s\n- heat-functional-wallaby https://zuul.opendev.org/t/openstack/build/b11c5eaad95a4b55879e715f75ceb817 : FAILURE in 1h 02m 33s\n- heat-functional-xena https://zuul.opendev.org/t/openstack/build/e3b298cfbd2d4b2b833cfed3b6af8b28 : FAILURE in 58m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c88447a2ca74c90f5fe8aa38ff6738f697feb620","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-18 23:46:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76e18d1ee9824955669e3b0e30f01f59e1921665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 00:44:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/496b91c8e05742e38338a03a54762b6f : SUCCESS in 3m 06s\n- heat-functional https://zuul.opendev.org/t/openstack/build/328ffc3e168b4ca6a071b5922ead1b45 : SUCCESS in 49m 16s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/fe12d778cae341ecbbbe034b8df9fd95 : FAILURE in 42m 14s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/5d024b6641e244608b3e0c654f405128 : FAILURE in 32m 53s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/5d428fe3b3424228b01ba4809b2b9879 : FAILURE in 33m 38s (non-voting)\n- heat-functional-ussuri https://zuul.opendev.org/t/openstack/build/475590bf04f0471187823110a13ee55a : SUCCESS in 55m 21s\n- heat-functional-victoria https://zuul.opendev.org/t/openstack/build/724a1730656f4dc0a3f7df2c76927802 : SUCCESS in 49m 50s\n- heat-functional-wallaby https://zuul.opendev.org/t/openstack/build/2527f9c9ffcf4460a5a837bb5b4200f1 : SUCCESS in 52m 33s\n- heat-functional-xena https://zuul.opendev.org/t/openstack/build/8a2d1ddbcd7941c4abc6feb624387c2b : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3efa116b36d30240104844dad1beadafd2292a5f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 02:53:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f7cf6a8452a370b9e3fec34cb8fed0b28739e33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 04:17:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8a57643c82646ac835b6ecb23e167c9 : SUCCESS in 4m 36s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c0c635950f2844b097833ce876d49d0d : SUCCESS in 46m 55s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d469b8defd394084866d33bc760e0bbf : SUCCESS in 39m 14s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/922f8e2e3a0d4a00b4bbcc6331fb64e3 : FAILURE in 48m 52s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/c5b3d0ad06ad424cb2aa83635875e492 : FAILURE in 49m 26s (non-voting)\n- heat-functional-ussuri https://zuul.opendev.org/t/openstack/build/4ace678b42544a6e9b79e50f0b4da9f0 : SUCCESS in 1h 02m 39s\n- heat-functional-victoria https://zuul.opendev.org/t/openstack/build/fa2ff924b4434a36ac6e97901f09bc88 : SUCCESS in 48m 50s\n- heat-functional-wallaby https://zuul.opendev.org/t/openstack/build/a1c7116ce5c846bfafea7878c61eab4d : SUCCESS in 50m 48s\n- heat-functional-xena https://zuul.opendev.org/t/openstack/build/945fd275c7754315a6f764e4808b5651 : FAILURE in 58m 22s","accounts_in_message":[],"_revision_number":1},{"id":"01775ac4d3b29a406b35cdf7ecf18ab7807e0695","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-09-16 09:13:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18af3c03021de1142593ab1727dd98a4c94c2d5d","revisions":{"18af3c03021de1142593ab1727dd98a4c94c2d5d":{"kind":"REWORK","_number":1,"created":"2022-01-07 12:51:02.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/94/823794/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/94/823794/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/94/823794/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/94/823794/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/94/823794/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/94/823794/1"}}},"commit":{"parents":[{"commit":"ba43685c131d6387779ed35f5bf61fc86667a728","subject":"Merge \"Revert \"Make all functional jobs non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/ba43685c131d6387779ed35f5bf61fc86667a728"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-07 12:48:04.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-07 12:50:46.000000000","tz":0},"subject":"DNM testing out the plugin","message":"DNM testing out the plugin\n\nChange-Id: I66de53e9f02d53b5c36f1b880c00106372ddb582\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/18af3c03021de1142593ab1727dd98a4c94c2d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/18af3c03021de1142593ab1727dd98a4c94c2d5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
