)]}'
{"id":"openstack%2Fheat-tempest-plugin~841136","triplet_id":"openstack%2Fheat-tempest-plugin~master~I1a44b65768cb862e10d013801dd47cb0510b2d07","project":"openstack/heat-tempest-plugin","branch":"master","topic":"remove-six","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-20 13:00:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I1a44b65768cb862e10d013801dd47cb0510b2d07","subject":"Remove six","status":"MERGED","created":"2022-05-09 15:51:40.000000000","updated":"2023-12-20 14:21:30.000000000","submitted":"2023-12-20 14:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841136-remove-six","meta_rev_id":"eef4c7636c6aae74a99b3ff5c4e290e81601ac64","_number":841136,"virtual_id_number":841136,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-20 14:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"all":[{"value":1,"date":"2023-04-17 13:12:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2023-12-20 13:00:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2023-12-20 13:00:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 16:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 13:12:06.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"}],"messages":[{"id":"8a873ad02a1cdd2edc73fb4e84fb5ec5760f8989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-09 15:51:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc06fd18b0c3f76c05a69f8476f6d17935e099f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-09 15:53:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"86fb4021e14eaead9be9600ab33abee6218e3163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 16:54:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937d04f330734f0d8b0fa84469d37d78 : SUCCESS in 3m 23s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c2090a764aee43b88af5058af94bc908 : SUCCESS in 57m 32s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/87326acf844f438da5b144f6cc935098 : SUCCESS in 49m 54s\n- heat-functional-convg-train https://zuul.opendev.org/t/openstack/build/13e16d4b0d924ebe8fac092a7a1045c7 : SUCCESS in 58m 26s\n- heat-functional-orig-train https://zuul.opendev.org/t/openstack/build/2abf5f98bf414d1a9dd6582cf5ccad75 : SUCCESS in 56m 27s (non-voting)\n- heat-functional-ussuri https://zuul.opendev.org/t/openstack/build/765fd90d85fb4c548a10f920312f51b0 : SUCCESS in 56m 05s\n- heat-functional-victoria https://zuul.opendev.org/t/openstack/build/5aa6dda96ac04b588db44a1aad07be27 : SUCCESS in 57m 12s\n- heat-functional-wallaby https://zuul.opendev.org/t/openstack/build/4fa68ad2ff894b78b4febe95c0ff5bc0 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"cf42bbd8ae89e6f00276c33129c4d48255149f6d","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2023-04-17 13:12:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7347b0a1bcc4b75220fc8ddc0795e5a01422785d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-20 13:00:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e08dce996c6e3db4e6948da6367350d2d342770a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 13:01:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8bcf8a355ce694c5ae2bd0ed707453a2f2b19c69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 14:21:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2205079a463b4982a17a6590c4f156ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfff4aa0241643de8b05e09bfc2c25d8 : SUCCESS in 2m 42s\n- heat-functional https://zuul.opendev.org/t/openstack/build/65d4e6e8a7f140b2a878ad89147c74f4 : SUCCESS in 1h 04m 26s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/6afce83b928e4ef49f2661b821f8be83 : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":2},{"id":"eef4c7636c6aae74a99b3ff5c4e290e81601ac64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 14:21:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfb9739a8d9d87ea52ba97345cee093c9d0db0b8","revisions":{"1bb471c10a5da50d90c044f61dae40bcb5632d35":{"kind":"REWORK","_number":1,"created":"2022-05-09 15:51:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/841136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/36/841136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/1"}}},"commit":{"parents":[{"commit":"ba43685c131d6387779ed35f5bf61fc86667a728","subject":"Merge \"Revert \"Make all functional jobs non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/ba43685c131d6387779ed35f5bf61fc86667a728"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-09 15:51:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-09 15:51:14.000000000","tz":540},"subject":"Remove six","message":"Remove six\n\nThis library no longer supports Python 2, thus usage of six can be\nremoved.\n\nChange-Id: I1a44b65768cb862e10d013801dd47cb0510b2d07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/1bb471c10a5da50d90c044f61dae40bcb5632d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/1bb471c10a5da50d90c044f61dae40bcb5632d35"}]},"branch":"refs/heads/master"},"dfb9739a8d9d87ea52ba97345cee093c9d0db0b8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-09 15:53:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/841136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/36/841136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/36/841136/2"}}},"commit":{"parents":[{"commit":"ba43685c131d6387779ed35f5bf61fc86667a728","subject":"Merge \"Revert \"Make all functional jobs non-voting\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/ba43685c131d6387779ed35f5bf61fc86667a728"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-09 15:51:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-09 15:53:21.000000000","tz":540},"subject":"Remove six","message":"Remove six\n\nPython 2 is no longer supported, thus usage of six can be removed.\n\nChange-Id: I1a44b65768cb862e10d013801dd47cb0510b2d07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/dfb9739a8d9d87ea52ba97345cee093c9d0db0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/dfb9739a8d9d87ea52ba97345cee093c9d0db0b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
