)]}'
{"id":"openstack%2Fheat-tempest-plugin~896184","triplet_id":"openstack%2Fheat-tempest-plugin~master~I9cf7532d51843698b63bae2ee825f683f07e2df8","project":"openstack/heat-tempest-plugin","branch":"master","topic":"lbaas-cleanup","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2023-12-16 17:58:01.000000000","reason":"Change was submitted"},"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2023-12-16 17:58:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9cf7532d51843698b63bae2ee825f683f07e2df8","subject":"Remove tests requiring Neutron LBaaS","status":"MERGED","created":"2023-09-22 06:04:10.000000000","updated":"2023-12-16 17:58:01.000000000","submitted":"2023-12-16 17:58:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896184-lbaas-cleanup","meta_rev_id":"82b54b8ffe3b2c8a07c4ed2a9a22ce078b4e9f44","_number":896184,"virtual_id_number":896184,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-16 17:58:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-12-16 16:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-12-16 16:36:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 06:09:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2023-09-22 06:09:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2023-09-22 07:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a981e25b191b02cfd64253a9ef9719bad7f3c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-22 06:04:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"949ad347c99091c01795d50195feea197f257175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-22 06:07:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"960439658835f7bc13d65ef7f6fa8ef5ecdcdaa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 07:21:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4d7e38d016480a83e642b1762381d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5747b012a5764b6d8f71bf9935f46c0d : SUCCESS in 3m 01s\n- heat-functional https://zuul.opendev.org/t/openstack/build/1e57cd1ae21342eba2bbb824ae965134 : SUCCESS in 57m 16s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d2e4135e22064c739ee645f21b3c646a : SUCCESS in 1h 06m 19s\n- heat-functional-2023-1 https://zuul.opendev.org/t/openstack/build/ba3ab4566e2e472f9acd924695585230 : SUCCESS in 42m 48s\n- heat-functional-zed https://zuul.opendev.org/t/openstack/build/8169dd1753e340df98df0fdab52b79ff : SUCCESS in 47m 39s\n- heat-functional-yoga https://zuul.opendev.org/t/openstack/build/8091acc471f3434e8742e583d272e99f : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":2},{"id":"61fb5d77604f64a36504003c9af231410ac154c3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-16 16:36:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"641c38944d695d6f5122a9e989e055d0f80fd961","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 16:37:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"17454e18b7d805b20a24a09ab40969c4a3063dc4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-16 16:46:22.000000000","message":"Topic set to lbaas-cleanup","accounts_in_message":[],"_revision_number":2},{"id":"f3673b8c979b017cb620d8be816b103816ad031a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 17:58:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1b3efcab014ee9af38490d813010cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5f3e549c7a4ccea9ff5e6d82637938 : SUCCESS in 2m 48s\n- heat-functional https://zuul.opendev.org/t/openstack/build/86e9f9ee4a8f48e8b42cb29ee7beea66 : SUCCESS in 1h 08m 07s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a0a89ac3eae943989ee281458237d401 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":2},{"id":"82b54b8ffe3b2c8a07c4ed2a9a22ce078b4e9f44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-16 17:58:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5047e210dca43b72b65361bd635ec6757844fc27","revisions":{"6b9d9daade108bd407ada08df21594b721de2672":{"kind":"REWORK","_number":1,"created":"2023-09-22 06:04:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/896184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/84/896184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/1"}}},"commit":{"parents":[{"commit":"26f59bd29a217f2b64f7a5b295eabc86db79ac46","subject":"Merge \"Remove some notes no longer valid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/26f59bd29a217f2b64f7a5b295eabc86db79ac46"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-22 06:01:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-22 06:03:58.000000000","tz":540},"subject":"Remove tests requiring Neutron LBaaS","message":"Remove tests requiring Neutron LBaaS\n\nNeutron LBaaS was retired during train cycle and was replaced by\noctavia. Because we no longer support train, it does not make any\nsense to maintain tests requiring that removed component.\n\nChange-Id: I9cf7532d51843698b63bae2ee825f683f07e2df8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/6b9d9daade108bd407ada08df21594b721de2672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/6b9d9daade108bd407ada08df21594b721de2672"}]},"branch":"refs/heads/master"},"5047e210dca43b72b65361bd635ec6757844fc27":{"kind":"REWORK","_number":2,"created":"2023-09-22 06:07:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/896184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-tempest-plugin","ref":"refs/changes/84/896184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-tempest-plugin refs/changes/84/896184/2"}}},"commit":{"parents":[{"commit":"26f59bd29a217f2b64f7a5b295eabc86db79ac46","subject":"Merge \"Remove some notes no longer valid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/26f59bd29a217f2b64f7a5b295eabc86db79ac46"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-22 06:01:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-22 06:07:17.000000000","tz":540},"subject":"Remove tests requiring Neutron LBaaS","message":"Remove tests requiring Neutron LBaaS\n\nNeutron LBaaS was retired during train cycle and was replaced by\noctavia. Because we no longer support train, it does not make any\nsense to maintain tests requiring that removed component.\n\nChange-Id: I9cf7532d51843698b63bae2ee825f683f07e2df8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/5047e210dca43b72b65361bd635ec6757844fc27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-tempest-plugin/commit/5047e210dca43b72b65361bd635ec6757844fc27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
