)]}'
{"id":"openstack%2Fheat-translator~654875","triplet_id":"openstack%2Fheat-translator~master~I55cf04680ac0c03eaf5dd66d3a2d7d5577220391","project":"openstack/heat-translator","branch":"master","hashtags":[],"change_id":"I55cf04680ac0c03eaf5dd66d3a2d7d5577220391","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 09:00:04.000000000","updated":"2019-06-12 13:00:51.000000000","submitted":"2019-06-12 13:00:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654875-1560344451061-41fe81b9","meta_rev_id":"d70987d13c134003d5f366985967ff86f2dc6ded","_number":654875,"virtual_id_number":654875,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-06-12 13:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"recommended":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"all":[{"value":1,"date":"2019-06-12 12:30:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 12:48:09.000000000","_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"all":[{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 12:48:09.000000000","_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 12:30:11.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-12 12:48:09.000000000","updated_by":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"reviewer":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"state":"REVIEWER"},{"updated":"2019-06-12 13:00:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7214a7a816fed145bbf1fa0d784359af02858a8","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 09:00:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d0b0cd07ad667d9e66aeb501840e14522bc670a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 09:12:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/75/654875/1/check/test-release-openstack/8cf7151/ : SUCCESS in 5m 27s\n- openstack-tox-cover http://logs.openstack.org/75/654875/1/check/openstack-tox-cover/f69675b/cover/ : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/654875/1/check/openstack-tox-lower-constraints/1976695/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/75/654875/1/check/openstack-tox-pep8/0b0d082/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/75/654875/1/check/openstack-tox-py27/5f7169c/ : FAILURE in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/75/654875/1/check/openstack-tox-py36/36bc7ae/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/75/654875/1/check/openstack-tox-py37/49b1f55/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/75/654875/1/check/openstack-tox-docs/0cdc216/html/ : SUCCESS in 3m 52s\n- heat-translator-tox-py27-tp http://logs.openstack.org/75/654875/1/check/heat-translator-tox-py27-tp/76679c7/ : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6756c888339b05abb31930fa3ee8d08b92155341","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-12 11:45:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a91a1139c4f7dd1bf1f9452adc1694e92855c26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:12:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/75/654875/1/check/test-release-openstack/7e2f774/ : SUCCESS in 3m 37s\n- openstack-tox-cover http://logs.openstack.org/75/654875/1/check/openstack-tox-cover/0198597/cover/ : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/654875/1/check/openstack-tox-lower-constraints/e1a3887/ : SUCCESS in 6m 16s\n- openstack-tox-pep8 http://logs.openstack.org/75/654875/1/check/openstack-tox-pep8/3b7c460/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/75/654875/1/check/openstack-tox-py27/bec3253/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/75/654875/1/check/openstack-tox-py36/2d731eb/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/75/654875/1/check/openstack-tox-py37/369c1a6/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/75/654875/1/check/openstack-tox-docs/d64a45e/html/ : SUCCESS in 7m 19s\n- heat-translator-tox-py27-tp http://logs.openstack.org/75/654875/1/check/heat-translator-tox-py27-tp/e31e007/ : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec15fe6f6dd132cb0fadefc572d47cf7b594cfce","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-12 12:30:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1f5043fc48caebdf73ea231adf44680856e3c096","author":{"_account_id":16511,"name":"Bob Haddleton","email":"bob.haddleton@nokia.com","username":"bobh"},"date":"2019-06-12 12:48:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"61862eff95b5c95ff2cf138637803300aa4ba9c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:48:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d621f8e2ebb83ce75697f4c71de4ddec365f6c6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 13:00:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/75/654875/1/gate/test-release-openstack/70ac1a7/ : SUCCESS in 3m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/654875/1/gate/openstack-tox-lower-constraints/607f755/ : SUCCESS in 4m 52s\n- openstack-tox-pep8 http://logs.openstack.org/75/654875/1/gate/openstack-tox-pep8/1e63c49/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/75/654875/1/gate/openstack-tox-py27/009daec/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/75/654875/1/gate/openstack-tox-py36/3e1aa93/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/75/654875/1/gate/openstack-tox-py37/fe99fd6/ : SUCCESS in 4m 38s\n- openstack-tox-docs http://logs.openstack.org/75/654875/1/gate/openstack-tox-docs/efa2f0c/html/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d70987d13c134003d5f366985967ff86f2dc6ded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 13:00:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa9f02d23af04c3cb59955308091a5f6794cf64f","revisions":{"fa9f02d23af04c3cb59955308091a5f6794cf64f":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:00:04.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/75/654875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat-translator","ref":"refs/changes/75/654875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat-translator refs/changes/75/654875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat-translator refs/changes/75/654875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat-translator refs/changes/75/654875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat-translator refs/changes/75/654875/1"}}},"commit":{"parents":[{"commit":"33ac9fd9b75408e12d3ddd7a3cd942dea106038c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-translator/commit/33ac9fd9b75408e12d3ddd7a3cd942dea106038c"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:59:45.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-23 08:59:45.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I55cf04680ac0c03eaf5dd66d3a2d7d5577220391\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-translator/commit/fa9f02d23af04c3cb59955308091a5f6794cf64f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat-translator/commit/fa9f02d23af04c3cb59955308091a5f6794cf64f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
