)]}'
{"id":"openstack%2Fheat~663404","triplet_id":"openstack%2Fheat~master~I2b02787bd8883ced631b81174cee9134445bf170","project":"openstack/heat","branch":"master","topic":"S2005523","hashtags":[],"change_id":"I2b02787bd8883ced631b81174cee9134445bf170","subject":"Support domain in keystone lookups","status":"MERGED","created":"2019-06-05 19:07:38.000000000","updated":"2019-10-10 13:09:03.000000000","submitted":"2019-10-08 18:51:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663404-1570560660661-70f2cce9","meta_rev_id":"0f35a3245297bc6ca8a72cf11a2a5a74e9a9a2b3","_number":663404,"virtual_id_number":663404,"owner":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"date":"2019-10-03 14:54:11.000000000","_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 18:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2019-10-01 13:57:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":2,"date":"2019-10-08 16:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},{"value":0,"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"value":1,"date":"2019-10-08 16:21:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 14:47:56.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-09-10 16:29:29.000000000","updated_by":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2019-10-01 13:57:45.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2019-10-08 16:21:31.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2019-10-08 18:51:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"15f0c74d6b61124a4d70fc3c45da6f1ffcb64208","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-06-05 19:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b83da80d771c2d329be17ea6216c44067ac50c49","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-06-05 19:18:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f230479290ccd4d48c5c35787c937c6190eca11f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 21:07:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/04/663404/2/check/openstack-tox-cover/7a38cbd/cover/ : SUCCESS in 9m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/663404/2/check/openstack-tox-lower-constraints/f8cf862/ : SUCCESS in 12m 28s\n- openstack-tox-pep8 http://logs.openstack.org/04/663404/2/check/openstack-tox-pep8/6d3e899/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/04/663404/2/check/openstack-tox-py27/ddc3a7c/ : SUCCESS in 7m 59s\n- openstack-tox-py36 http://logs.openstack.org/04/663404/2/check/openstack-tox-py36/2e43d34/ : SUCCESS in 7m 10s\n- openstack-tox-py37 http://logs.openstack.org/04/663404/2/check/openstack-tox-py37/3c870d7/ : SUCCESS in 7m 47s\n- openstack-tox-docs http://logs.openstack.org/04/663404/2/check/openstack-tox-docs/523842e/html/ : SUCCESS in 6m 07s\n- grenade-heat http://logs.openstack.org/04/663404/2/check/grenade-heat/d0596dd/ : SUCCESS in 1h 13m 52s\n- grenade-heat-multinode http://logs.openstack.org/04/663404/2/check/grenade-heat-multinode/4847d9d/ : SUCCESS in 1h 26m 00s\n- heat-functional-orig-mysql-lbaasv2 http://logs.openstack.org/04/663404/2/check/heat-functional-orig-mysql-lbaasv2/b4d2930/ : SUCCESS in 1h 28m 41s\n- heat-functional-convg-mysql-lbaasv2 http://logs.openstack.org/04/663404/2/check/heat-functional-convg-mysql-lbaasv2/0d141f2/ : SUCCESS in 1h 24m 39s\n- heat-functional-convg-mysql-lbaasv2-amqp1 http://logs.openstack.org/04/663404/2/check/heat-functional-convg-mysql-lbaasv2-amqp1/fbbe4de/ : SUCCESS in 1h 26m 11s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache http://logs.openstack.org/04/663404/2/check/heat-functional-convg-mysql-lbaasv2-non-apache/680eabe/ : SUCCESS in 1h 22m 36s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 http://logs.openstack.org/04/663404/2/check/heat-functional-convg-mysql-lbaasv2-py3/46b247a/ : SUCCESS in 1h 11m 45s","accounts_in_message":[],"_revision_number":2},{"id":"fbdfb4770220db1d19880864b60ca859e0ed5476","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-06-10 13:49:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nWould be nice to have a unit test for the new functionality as well.","accounts_in_message":[],"_revision_number":2},{"id":"ba6e8b68329c094d41860f7381e8a2b81a1a4c91","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-08-09 20:00:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7eaab3dd46876d88b22fe872421f4e491a83c570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 22:28:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://logs.opendev.org/04/663404/3/check/openstack-tox-cover/7ad130c/cover/ : SUCCESS in 10m 31s\n- openstack-tox-lower-constraints https://logs.opendev.org/04/663404/3/check/openstack-tox-lower-constraints/7a896f6/ : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://logs.opendev.org/04/663404/3/check/openstack-tox-pep8/53738ba/ : FAILURE in 5m 01s\n- openstack-tox-py27 https://logs.opendev.org/04/663404/3/check/openstack-tox-py27/bc8f68b/ : SUCCESS in 7m 35s\n- openstack-tox-py36 https://logs.opendev.org/04/663404/3/check/openstack-tox-py36/1043e11/ : SUCCESS in 11m 24s\n- openstack-tox-py37 https://logs.opendev.org/04/663404/3/check/openstack-tox-py37/6b61617/ : SUCCESS in 9m 43s\n- openstack-tox-docs https://logs.opendev.org/04/663404/3/check/openstack-tox-docs/ad1571a/html/ : SUCCESS in 6m 40s\n- grenade-heat https://logs.opendev.org/04/663404/3/check/grenade-heat/902a025/ : SUCCESS in 1h 10m 20s\n- grenade-heat-multinode https://logs.opendev.org/04/663404/3/check/grenade-heat-multinode/ecfd730/ : SUCCESS in 1h 19m 20s\n- heat-functional-orig-mysql-lbaasv2 https://logs.opendev.org/04/663404/3/check/heat-functional-orig-mysql-lbaasv2/9929a46/ : SUCCESS in 1h 23m 42s\n- heat-functional-convg-mysql-lbaasv2 https://logs.opendev.org/04/663404/3/check/heat-functional-convg-mysql-lbaasv2/3aa1936/ : SUCCESS in 1h 35m 31s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://logs.opendev.org/04/663404/3/check/heat-functional-convg-mysql-lbaasv2-amqp1/d21fcc6/ : SUCCESS in 1h 31m 04s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://logs.opendev.org/04/663404/3/check/heat-functional-convg-mysql-lbaasv2-non-apache/23c322c/ : SUCCESS in 1h 26m 40s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://logs.opendev.org/04/663404/3/check/heat-functional-convg-mysql-lbaasv2-py3/f752dda/ : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":3},{"id":"1dbc105bf7a9c8f05c72c8ee510a5fabde056d85","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-08-12 14:42:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0b5173e198b2c5a84c52cbf93243365b518b4c4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 16:47:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d58801cc713148a69b9f9bf73251699e : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df044f94186a424eb267ef82a8562ca8 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/054212ab2ba74b95a3562fff2d5f06d7 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cea5946a7ee3487ca8e70c65c30b8919 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e392e12587947169040093399cbfda4 : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00ae578c30764d6fb50cc0e55b0f3e8f : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256c08fef2ce47bf9411b815d8ec2aef : SUCCESS in 6m 18s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/4abcbe9c44444fd79fac345042911e8a : SUCCESS in 1h 16m 18s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1c6e6a1cbcce400f9eee6398e4374920 : SUCCESS in 1h 30m 14s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/fda0405b4d9940398c4ae2ed1d09457c : SUCCESS in 1h 43m 55s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/58bdfd7bedd140dabe2de1a983560f79 : SUCCESS in 1h 38m 06s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/f26665443e32497e9c62e4f29cb67191 : SUCCESS in 1h 34m 51s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/8b659281f5ec48108884deee28d70995 : SUCCESS in 1h 31m 39s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/cdbd9461a19e4974ab290a88a4650ff3 : SUCCESS in 1h 09m 45s","accounts_in_message":[],"_revision_number":4},{"id":"cbe45299166b7effd290d171afe62865c1da73ae","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-08-12 17:12:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6e41cf5d3f989a42e4ad7ffae299a118ac3862ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 18:59:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfa4a65e528649d0a2d9ba0d4054b371 : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/601101658ea2446a91efd16b08ed7f79 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07ceaf6f2cf499db6fbd01247fbfd41 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02b7dd33d3b64f7b93053f2fc2e20078 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/799a8f84194d46719a0d76832c1c01b6 : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/706f9cc4e92449688fcbe99e75054ded : SUCCESS in 9m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec969bf2231b4dd79722886ba3413259 : SUCCESS in 7m 41s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d1c8c9dcb85b43a78ac4f562eee1357f : SUCCESS in 1h 09m 51s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/fbc547cfdd57456d915353aab5aa915d : FAILURE in 1h 22m 40s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/0696e28ad20c423aa43eaa168b6c40cc : SUCCESS in 1h 18m 43s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/092270895f9449138368c98f6857c308 : SUCCESS in 1h 36m 32s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/df0511da6bf24b5cb396524c1b7cbe59 : SUCCESS in 1h 29m 25s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/3dd5bcf94966440399556e7b3a752294 : SUCCESS in 1h 36m 18s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/084e04d761d847bd809a785504b01208 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":5},{"id":"c8cc82aaca85389ebf9125bb9632e9c620a87c4d","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-08-13 14:47:56.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2a3b220f7930262ef0863ab138b8e54b94b3a699","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-08-13 15:23:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5a5d2306ac251c97066ec169f164009af1f087bd","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-08-13 15:30:20.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"8deb37d07c8ecb484fb6c93ecb190d6a3cc3a0dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 17:41:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/833cb1f20b4e4917ac767663ac5b2b76 : SUCCESS in 9m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/855f34473f0f4d29bf685e506c3e537d : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164447c864374e1fa24ac7101e36cc74 : FAILURE in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f4db5628cdb849ee96be1d3b8acc9048 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa218daf08854300b9a9ac6b5fbd77b5 : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca4373152b5f49cdaa298a2c3db6f1a1 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20d179b3a644949b1a1a6855f059dc3 : SUCCESS in 8m 41s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/bf474195b0f444788c7ab2f215d9472d : SUCCESS in 1h 27m 26s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/863b074b014e45cebfbe1bbf6d98362c : SUCCESS in 1h 39m 08s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/87248f809b344a5b951e2c8acecaf06d : SUCCESS in 1h 32m 20s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/1f7f677825c1440c8381ee8e2faae81f : FAILURE in 1h 42m 13s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/4f352a59a7ea456a8e658a7b00d468ce : FAILURE in 2h 08m 15s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/82cfafeee1df4decb001ab15c2b97ece : FAILURE in 1h 42m 08s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/f5f4b33806b7496594dae03da8fee8eb : SUCCESS in 1h 13m 37s","accounts_in_message":[],"_revision_number":7},{"id":"bf404bd9953ea5b9ff98ca92b85602037c1b7fe8","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2019-08-13 17:51:04.000000000","message":"Patch Set 7: Code-Review+1\n\nThis looks good on my read through.","accounts_in_message":[],"_revision_number":7},{"id":"adc07ea7583ea1bb3eb668277cbf1c163644de34","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-08-13 18:24:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"14173a6c1baf48d171209c2aa94bbaa668d5ed7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 20:16:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e86ba11acc2641c4b6ccdfdf58334c9e : SUCCESS in 9m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bce5d14db16a48bf872d26a5ecfd57e0 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4182a75df1dc4a2da3f3dd5fde71b3bb : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e203d582c38d43978287c2e180087daf : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a39a70931d1435db5a7b46df0e545bf : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c4b7149e49a4a98bb8e1723d41cd9a4 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5874e92ec02f42c4979d0ac3d883c317 : SUCCESS in 8m 04s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/03dfc7da8fa74f13892b70aafe597044 : SUCCESS in 1h 15m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/f9fedac3a83c49549719305189bd5d1f : SUCCESS in 1h 20m 03s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/7384c730df514f798f13bfbfae08abd3 : SUCCESS in 1h 16m 38s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/d036d2e53670494ba87d04ebe462ea64 : SUCCESS in 1h 26m 37s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/85c69db9af1349119708ae768c0f1e64 : SUCCESS in 1h 32m 16s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/4975d7a275294d4bab48eaf2428294e2 : SUCCESS in 1h 42m 05s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/13bec5a960f746b299edee394976cbb0 : SUCCESS in 1h 26m 32s","accounts_in_message":[],"_revision_number":8},{"id":"e2b272ff7ded987f5fbafa075cf27b999ac30312","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-08-14 13:40:48.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)\n\nThis looks like a valid approach, just some minor comments on implementation.\n\nPlease also add a release note in this commit (see https://docs.openstack.org/reno/latest/user/usage.html#creating-new-release-notes).","accounts_in_message":[],"_revision_number":8},{"id":"b338637af6da8702fc0dc1631ce1563dc5f67b89","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-08-14 13:46:55.000000000","message":"Patch Set 8:\n\n(3 comments)\n\nsome other comment along with zane\u0027s. And I think we can use a little more info in commit message so it\u0027s more clear what\u0027s User and group search across domains or just adding example in Story?:)","accounts_in_message":[],"_revision_number":8},{"id":"a49f9d047b4212e50a36596285d1675bbba33101","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-08-15 13:44:00.000000000","message":"Patch Set 8:\n\n(6 comments)\n\nUpdated comments","accounts_in_message":[],"_revision_number":8},{"id":"7446933cac6320c3c01d9d5aed4371c9c430d7d3","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-04 15:21:11.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"0cf788f1993401b91cb829caa13c654674cff397","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:50:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cecdf1a406754155ae704744d77709f9 : SUCCESS in 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/184f71eb46324932b8dd75f99e8d8ccd : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffe4cdbe5d0449185acb3eb06fb2b77 : POST_FAILURE in 7m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6814b3de22f341aeb797f71ed10740e4 : SUCCESS in 8m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/136b7fc14a0d496d92e29b73db7a9f82 : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85cf5c10e62c49d9bd4325858be22139 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51b5c29578a84187b4a9444b75eea399 : SUCCESS in 7m 57s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/9ec53f100c184bd980e7e460f34e7103 : SUCCESS in 1h 20m 12s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1327427915d04606b7353f7f486e3629 : FAILURE in 48m 46s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/c93ed194a6364f368d7be6eda38ec52b : SUCCESS in 1h 29m 15s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/2925bfb24c374d1999aa82d7a58bf9e2 : SUCCESS in 1h 26m 31s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/ad61d882167d46d5b8536ee132e1b9a7 : POST_FAILURE in 1h 24m 45s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/94ebf7ad6bc04346a240d8d82e6029a9 : POST_FAILURE in 1h 35m 04s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/02349065ed8d42169c25bdbbea6db638 : POST_FAILURE in 1h 21m 19s","accounts_in_message":[],"_revision_number":9},{"id":"bf5218401c1a37bbdc9e3dec641d5a8f4ae737e2","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-05 14:02:15.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e1f018b76e3b0f1c36849739a2b0cf5c316dade0","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-05 18:23:15.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6e076bbd792ddb11837ac20776b0cd42dce88e7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 20:12:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3618c0b6e704f7eb1d48a343f9bbcb6 : SUCCESS in 11m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5feffd6cc014d0ba72f84c63b2a91f8 : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a484235fa3411298f3475ca6749051 : SUCCESS in 7m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d22b0034bf1e4a17991d3fbcd73dae57 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edb51b70d79f4e9b97c69a6130eed778 : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/473c5d20fa724ddbbdbcbc1cc0ad0648 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64fc0f9d7bec42708cc3794ed07a6a0b : SUCCESS in 16m 07s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/2bc8591aab19490ca6746f12a5266118 : SUCCESS in 1h 18m 29s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c80cbfdaa4be40f78c8b51f1c48eb2b7 : SUCCESS in 1h 24m 02s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/f7edb6dc911943bd969975580e053a24 : SUCCESS in 1h 17m 56s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/ccde5d5ae19e4c8e9b38ac1406e13465 : SUCCESS in 1h 21m 59s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/d2848477736f4dcdb377ded1fc1317d6 : SUCCESS in 1h 37m 09s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/87221ccc111e4b4ab6976819e8214cc0 : SUCCESS in 1h 30m 53s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/23d08e13e06b444f962b4514a7f610bd : FAILURE in 1h 10m 46s","accounts_in_message":[],"_revision_number":10},{"id":"6d5125d6b99b688b602a5e913c1d71fc764a7f3a","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-06 13:00:32.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"b90575b31ce2385dbec88ea48b32b380328be214","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:49:32.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14a0f7ae7c9445f7871e03ad057dbfea : SUCCESS in 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2951a6c9df454535940cdf4d6de3dcd6 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f997a5b8dce2462ebe89590a3a4939fe : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7b0c13882504f9583f716c7025a3c72 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d70c06fd269d4649af9ca7bfc72f963b : SUCCESS in 7m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/674f56d26b4f441ab9ae9ca611b35fe7 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2719d144fb664be1a64b09cb612195fd : SUCCESS in 7m 30s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/945447026d1641eea131056653d9734d : SUCCESS in 1h 18m 04s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/d720f9139af24599a82ff25c663174d7 : SUCCESS in 1h 33m 00s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/5b59050d42cb4c31a209f613d08ecced : SUCCESS in 1h 25m 58s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/bc228731978149b68fd4cd1b39718b1f : FAILURE in 1h 35m 30s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/2f71a7e1aa444ccdbb34e0d9ab56292b : FAILURE in 1h 45m 02s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/f0c578f0883f42499f4f54a413ced851 : SUCCESS in 1h 32m 11s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/86c6fdd7151b44b1bc871b94f4a6d079 : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":10},{"id":"1b8ec08729d4cfe4fedc3181cc4e6ec8e37ca14a","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-09 13:33:01.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"7718734e9613f5d16e1ba3273c7c2d233a64bc17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 15:20:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/277ae639325443feb8fe2ab88219f1ba : SUCCESS in 10m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd203132627e4d26bba9ebbc0cd5c196 : SUCCESS in 22m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb1a923560c4473a99c2633fdfca3a8 : SUCCESS in 10m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d16587e8762b4c73bfae99a19977421b : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ee09c355434f5fbd4c7152467803ae : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f74a26146fa84fd883cc229be9f7464b : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97fb0f47a6d641a08acc46904925b542 : SUCCESS in 10m 47s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/06009947137d4b2dbb1f09c7bdec5f8a : SUCCESS in 1h 30m 30s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/5fdf663cf47a4f128a8bd442a0a8bb8d : SUCCESS in 1h 37m 17s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/30dd2ab59d3f428c8a5bdce0f8214cce : SUCCESS in 1h 28m 00s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/a0a9930639b34cb2b163dfe074105802 : SUCCESS in 1h 25m 58s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/a24a3c0d521845bbaa9abd9ffc9bdbd8 : SUCCESS in 1h 32m 48s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/5c9cbbbf151e46b0a6001be248176661 : FAILURE in 1h 43m 15s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/825a9d424f354f0fba7b79ebcaace9eb : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":10},{"id":"7b4260a9220e4ff9df9220f1f2c2b9990307df23","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2019-09-09 18:02:19.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ba6a1feed2e40dd7ba4d8e32fd9521f66c3a8cf0","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-09 18:53:33.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d772e213073958a257a408a28253486984fac62b","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-09 18:59:59.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nIt is a typo, I updated the function name to parse_","accounts_in_message":[],"_revision_number":11},{"id":"0f8dcbab185a7209015d63f11cb5ca721a3660f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 20:48:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d312c0a2df44ce4829267ea746eed4a : SUCCESS in 9m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aaf7039bb044fb88e81cb0e2cfc83e6 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91d1b26b5a647c991bb8bc2677dafb8 : SUCCESS in 13m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4db7e018621f44a58c662dd4a79a00e7 : SUCCESS in 14m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/994dd76382924d5fbb0ee8f379701dfd : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a95da85fa75f41daa1af78c937867434 : SUCCESS in 9m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae75b82cab634528af281df70f53943f : SUCCESS in 7m 52s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/7d0988132c0b45b699dea43ad621b235 : SUCCESS in 1h 15m 59s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6798e2ee19754faeb1d7b58dc3dea272 : FAILURE in 1h 46m 06s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/bed7a5b2568641248676e6f1023a4bfd : SUCCESS in 1h 41m 42s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/a4b91085dc434d99871a9522743bb008 : SUCCESS in 1h 42m 00s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/be9b654597804689b49894ebbbe28c44 : SUCCESS in 1h 27m 46s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/494265a33f03464da7437d5c9fd4d634 : SUCCESS in 1h 26m 22s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/8b7e1303b763435e872c7248ea4dc43b : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":11},{"id":"0a71bbc3608e8e282c6c32b626b365c2467e818e","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-10 12:23:17.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"9986237211df596e5705bc38d5c0f2acd9dfaf2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:10:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f68a4d571ec4d93bbbf07d7b0c363cb : SUCCESS in 12m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8962e4f847054ebe888a805b60d86ca0 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176597db5da34ebe97269e3cb426ca4d : SUCCESS in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3cb4941a1a144240bed73e23f84fb0e8 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d0bd4c374b9491389ec71ee98f4be82 : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f2107e53f094782ba23730bbd662190 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f72b677c8e4965bf6180b041fd256c : SUCCESS in 7m 40s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/62d70478198846108c20a1474ed0b22e : SUCCESS in 1h 26m 45s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/04afdcf3cd0b4f1bbce1b9dabbeae1a2 : SUCCESS in 1h 32m 04s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/b6060f8cbccb4df888b112e69ac95670 : SUCCESS in 1h 27m 29s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/4ccff5536b5649d9a3f38f8d06f57928 : SUCCESS in 1h 29m 04s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/b68cee7876bb475eb6367f20391334f5 : SUCCESS in 1h 35m 44s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/329438f82cb2468e922bd36dfe41a887 : SUCCESS in 1h 25m 58s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/38462ff596b8419abe83bf37123c4e83 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":11},{"id":"c2a7c158a0eec01d558bcf92c19f17e3e6b95e9c","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2019-09-10 16:29:29.000000000","message":"Patch Set 11: Code-Review+1\n\nThanks Sam; looks good.","accounts_in_message":[],"_revision_number":11},{"id":"30435b8d437489f6b0c40cb2b997ed3fb8820b60","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-09-11 15:30:19.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"7a781a32ec3a998e578387b4e4c2f8133b905b70","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-09-11 16:26:22.000000000","message":"Patch Set 11: Code-Review+1\n\n(3 comments)\n\nOne bug in there, albeit in code that\u0027s not currently called. A couple of opportunities for simplification exist once that\u0027s fixed.\n\nPlease also add a release note.\n\nI\u0027m going to ask keystone folks to sanity-check this approach as well. (I like it.)","accounts_in_message":[],"_revision_number":11},{"id":"f25e1fdae8f8a2e06cb781f3d5fdfade4cac2837","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-09-11 17:13:48.000000000","message":"Patch Set 11: Code-Review-1\n\nOK, apologies for not doing this sooner, but I checked with the Keystone team and unfortunately there\u0027s a problem with this approach that we\u0027ll have to work around:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-keystone/%23openstack-keystone.2019-09-11.log.html#t2019-09-11T16:28:41\n\nBasically any character (including @) is allowed in a user/group/role/project/domain name. And in particular it\u0027s not only possible but extremely likely that there are people out there using email addresses as user names.\n\nI think we have an @ in the domain covered at this stage, because we\u0027re only splitting on the first @.\n\nWe could handle an @ in the username when no domain is specified by looking up the domain, and if that fails falling back to looking up the whole username string. That still might fail when only the username is specified if e.g. the username is u@example.com and there is an example.com domain; the way to fix that would be to fall back to the original string even if the domain is found but the user is found.\n\nTo handle an email-style username appearing with a domain then we\u0027d need to try splitting on the first @, and if that fails split on the second.\n\nAlternatively, we could try using a different and less likely character [sequence] as the delimiter. Puppet-openstack uses \"user::domain\" (mainly because that\u0027s common for puppet though). Maybe \"user|domain\"? Or \"user{domain}\"? I quite like that last one actually.\n\nThe only truly bulletproof way, and what the keystone team recommend, would be to add separate domain properties. However, I still think allowing domains in existing fields is on balance the most practical approach, since there\u0027s no way to do validation that requires data from multiple properties/parameters.","accounts_in_message":[],"_revision_number":11},{"id":"3a5b4e06716ae4254bfb7509cf42ae6042f7d5f0","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-12 13:03:11.000000000","message":"Patch Set 11:\n\n@Zane Bitter, thanks for these updates, good points. I\u0027ll update the implementation. Will ask any questions in IRC","accounts_in_message":[],"_revision_number":11},{"id":"3a8fa2098db9fa8e8c7da5bbdb9c4c5e30f95fa7","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-09-14 01:08:36.000000000","message":"Patch Set 11: -Code-Review\n\nRemove the+2 to honor Zane’s comments","accounts_in_message":[],"_revision_number":11},{"id":"002054a67d9b8cfe5b4f0bfc86cc36c3c63361ef","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-17 15:35:08.000000000","message":"Patch Set 11:\n\nI will implement  \"user{domain}\" that seems to be safest and easy to implement","accounts_in_message":[],"_revision_number":11},{"id":"02730f23c7f8f1ed34ae1089552e9b7c09ce9d02","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-19 18:34:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f58d3e563809a43a6b53050ecbefe1cfc6e9204d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 20:29:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b15d7afc5ca4da09a747ee162916e8d : FAILURE in 11m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5036927a94f4d8e8eb1e6b77d8e7e70 : FAILURE in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d8b94032104c6b9c479c04d6d57f71 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb25224a29ef49e09937883a081f9e6b : FAILURE in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebd246db1c34419ea9eadc9f3d9d7f55 : FAILURE in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0412910f9d546cfa99ca93700cae95d : FAILURE in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058b117153894d75975abf468bc5ca22 : SUCCESS in 8m 19s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/e96401463809441388595576ef0080fb : SUCCESS in 1h 22m 34s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c77d319c1eda424b8bc6e99abc298089 : SUCCESS in 1h 25m 57s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/b03e76d30cc940239d1e3bff387583b3 : SUCCESS in 1h 39m 00s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/fafcb9d0011e4cbea0aef4c4a1de475c : SUCCESS in 1h 38m 12s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/e4e302c11cdf4119853747a9190e08b6 : SUCCESS in 1h 36m 03s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/3d510c96889341e7b97babbbbd65b74e : SUCCESS in 1h 34m 12s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/08af0403cb09414f99b10eb92ab5882f : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":12},{"id":"d4f4fa53bc45bdadfc27bafffebc829e698d5f15","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-09-20 06:31:32.000000000","message":"Patch Set 12:\n\nyour test didn\u0027t update accordingly, also please kindly add releasenote for this change","accounts_in_message":[],"_revision_number":12},{"id":"1e0648580068daf614d3079d782514b3bbee5160","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-23 14:17:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ca09c3314ed44b0688d3677eed03879eeb2c8399","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-23 14:20:43.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d86f513d6a9ed860f5c9f371dacda9bfb7c0ef39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 16:10:39.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/017e1a067ed941b79094823d99c73fa3 : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e787892c598f45618fb78f5cc1fd80ec : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d21b1f3586a4ebc8d1edda33cb15772 : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ba4c4d0028a4760b7cc00d81c5e38d2 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/326761135979422f89b0928a26a11ad3 : SUCCESS in 8m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/341a58a816cf40b68cdcd0ed5735da07 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22db020bc6de4112b26185eb77ab270a : TIMED_OUT in 31m 11s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/2203b001b8494dcf801df3dd3f40af6b : FAILURE in 1h 37m 15s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c4bcdd0b8dab478c8429ee7d6056d761 : SUCCESS in 1h 42m 14s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/0329fb3645464c8982d8e921bc31b70f : FAILURE in 1h 30m 54s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/565158c6f55246f4a758d7f30e0b19e4 : SUCCESS in 1h 31m 22s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/fb4f21f1b95c4adbad41d0333a89c864 : SUCCESS in 1h 39m 58s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/9cf43a4f48c84a97a77b02927c676047 : SUCCESS in 1h 35m 22s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/93561af0fc0d41b79e6201325ef678e8 : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":14},{"id":"d2dbb0602d52e231196bfdb5847e3e2520187e23","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-24 12:27:17.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"7dc61b8508d8b3c5a6e8c74784a32f9feaa591b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:45:43.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed71aded95f14a3a9b70dbd70dddb7c8 : SUCCESS in 9m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d21bbfc84d7c4e71a9b296d445bf7708 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6af230d322047a4867bf72cdf362c53 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b3d6391a5c84193989a3085bffa76bd : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/209c4b54f7aa4ede9bdd4f38384436b9 : SUCCESS in 8m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef2f552413d04613ba8869e5e06bcb56 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93402b44b1f4484fa9fe61a35ffa46c1 : SUCCESS in 10m 05s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/c48978632eaf4c5eb6e93bccb44c982d : SUCCESS in 1h 15m 57s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4137c1d2210f4cbf8ca3c6a590c2d297 : SUCCESS in 1h 33m 24s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/af7e9b0e96f04a69a0ed18c6ab38343d : SUCCESS in 1h 23m 46s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/de56e5a536ef4fae811bf18518b66e93 : FAILURE in 2h 04m 31s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/a137b0e0140246148f5d86201dd52265 : FAILURE in 1h 43m 35s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/66f580be2f984ba8a636c10ab8f10a2d : SUCCESS in 1h 38m 29s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/3bf5cee5957143448c521a12717b9fc4 : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":14},{"id":"25a90243eac170c034367d9fa82397e6610ae847","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-24 19:33:32.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"67a798254def7149e65403a2c049689dae22488b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 21:28:54.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b820729919744d4e936d17750f3e3d02 : SUCCESS in 11m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1159a497e9a4c1d87d29aea7dee5ee8 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9552aa91a724eabb3c4972cfa3b8a7b : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45c1a66f9acb458baa276c7419393536 : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/604dc0f3bfd943ac800ed61a9dbc86fc : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d3bd79a934d4b06bd4d12284891b35e : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1d31bf0a284fefb1a67258743de231 : SUCCESS in 9m 16s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/b5e7da543f1b4a68a84b0dfa94d5dda9 : SUCCESS in 1h 19m 18s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/2cb9c00fd237431d9c22c5460afdeccd : SUCCESS in 1h 23m 02s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/0280f5e50739488a9ae86b1f6bc69272 : SUCCESS in 1h 16m 24s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/472d58c8a1964fc78fc26ce2812f9df5 : SUCCESS in 1h 48m 18s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/8fe2960d6af447b3b72d4c0891193c0d : SUCCESS in 1h 28m 08s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/e6b46e35100649699a905ad9c5cbd707 : SUCCESS in 1h 36m 52s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/5ce5e4d7e4d64080b3f358bbb01fb6d6 : SUCCESS in 1h 27m 31s","accounts_in_message":[],"_revision_number":14},{"id":"ab727ac314a2b312fd8873dda574d8600182e5cd","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-25 14:00:28.000000000","message":"Patch Set 14:\n\nWorking on the release note","accounts_in_message":[],"_revision_number":14},{"id":"205e794aeb179d2005cb96fe82e3931e63c89e58","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-09-25 15:00:28.000000000","message":"Patch Set 14:\n\nRelease note added here: https://review.opendev.org/#/c/684766/","accounts_in_message":[],"_revision_number":14},{"id":"05feabd58b87cdea3cf0c071d68f3d1e7521c186","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-09-27 17:33:59.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)\n\nIt would be better if you could squash the release notes into this same commit.","accounts_in_message":[],"_revision_number":14},{"id":"109d17c224f4c9c497f3f7fd65b41e39750b1f4d","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-10-01 13:38:58.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ade2d7560fb834cbb981c818b766010d6bf9ab26","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-10-01 13:42:53.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"bc1175a3d878e4b7c28429336c39c0c6bdeda30e","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-10-01 13:57:45.000000000","message":"Patch Set 16: Code-Review+2\n\nThanks! This LGTM.","accounts_in_message":[],"_revision_number":16},{"id":"d8e5730b7fe3dd142afb1348f934ee3e61d44b26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 15:51:26.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54368ef1950643bfa23fb01e000c1630 : SUCCESS in 9m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c025d988e0a949afb790c739a749ffe9 : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ddf2edfa4e48ff9d7e30b7be2b5774 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c59dd484a7804d758b5b21caf2ef9df3 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fad059fc33d7492eb425d684791c8017 : SUCCESS in 10m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b62d21ba0856410691bb394356f48837 : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1cf1a329ac42dc9543651223bf8d35 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1def03b6e96145188b83cb1c9224aaad : SUCCESS in 16m 58s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d7fc264fcc70478881977b2c1cf86e4a : FAILURE in 2h 01m 06s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/377ba7e12dad4f6e960ecf731ccf2153 : SUCCESS in 1h 45m 57s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/0da793cd53de400aa255ad10aa0bbe51 : SUCCESS in 1h 52m 27s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/4de27456d40343aab4d533b51d78d578 : SUCCESS in 1h 34m 20s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/a52e93b8ea994b08b8f15a6d6c94bca1 : SUCCESS in 1h 31m 14s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/b90239aeac5d461581131739081dbf48 : SUCCESS in 1h 28m 31s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/bc69fd5d74024a2b82b18156276e59ea : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":16},{"id":"69d4afad8eee9dbd5f0ef0b6d7f11a39c320ab99","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-10-01 16:10:57.000000000","message":"Patch Set 16:\n\n(2 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"e1059e319f2c8ed7c21a2ff5d2ce393aeca52394","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-10-02 12:59:54.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"4ab0b44ef4fc53ca0f150fad9579556ca13d9e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 14:46:25.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b39ab23aa244005ae196ea1e9b548e1 : SUCCESS in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/290d2293267744e7a4c92209d5a397f9 : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ca6812479b349338687785cb43155fb : SUCCESS in 6m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b8f45c8f200472a9d4151dc88fede16 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e800216a69d44b6ba2e3cd04ef65b050 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a54e22f349f34e189a75651de6193aed : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e916aa56e6b404d9b48ae170dfbde34 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0f1ef367c4e4d00b65c1a80e11ba58f : SUCCESS in 16m 38s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/1efd3f5f9da949759c59f08f8ac3ed29 : FAILURE in 36m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/5f2d0c91a5e34d28baaf754586841e4e : SUCCESS in 1h 10m 21s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/c1f924c6c10148e9a41d7c8b01f6213c : SUCCESS in 1h 21m 40s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/8783d429fb5642638e4a982500f07af3 : FAILURE in 1h 41m 32s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/7daebf2980f948ce8e97fe4fe2383ab2 : SUCCESS in 1h 44m 52s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/ccc9e14e73724484a4b17de635a14341 : SUCCESS in 1h 28m 51s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/63c9cea946474870a515301728991451 : SUCCESS in 1h 15m 53s","accounts_in_message":[],"_revision_number":16},{"id":"4799df51032601421a57b1f0ef86b0d221d7c8d9","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-10-02 15:32:30.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"81409797238c9c692880a011ccdf310d1a92ae7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 17:31:53.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ec76dcf94994527b2f620fd801e97a2 : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ee555f309224827a26b99aa2810f412 : SUCCESS in 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c9556f5651458db3510a851faf3033 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba267d18c0d6478b8992900ab44188f3 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2068574483d94ced8e95e99c63de279b : SUCCESS in 8m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/644fd9855f2346d6ad9fc9fd3f66c72a : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386869bebf594174825001344acbe606 : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4238dc50d25e43009aac8dd2df672a21 : SUCCESS in 19m 08s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/3801ee67d2a74e458e6311498ccdb3a3 : SUCCESS in 1h 18m 36s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/8eca15affc1d4b26897dd5e84a21d9e6 : FAILURE in 1h 51m 42s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/70d0117fdf03413cb72d364291ae4f2d : SUCCESS in 1h 25m 51s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/ca5856b5ede845a483f794ff347d9ac9 : SUCCESS in 1h 39m 15s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/b1e6c6a174964d509ce94f4c76fabf4b : FAILURE in 1h 44m 49s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/e41dfd81548a431fa6f5cd12196470b6 : SUCCESS in 1h 15m 31s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/4ca415ecaeb74ebd8df8429439c4b612 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":16},{"id":"0c51df31dcef7f04a2341a43bea8a190e0ecbf3b","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-10-03 14:54:11.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"d3bb8f912aad1bd25020adc874de8ad4f8de935d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 16:39:03.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c6aec7d4b41426eae1fc80e5e18bb62 : SUCCESS in 8m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fda9d11749c34047b6344e85bca68d06 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/120cb1cd0ae9484d80f840dd21844444 : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3592f4e0d6e545d0918a48ec4c710930 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bad5961b82e84253bd835df7c28e788a : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa88e7ad72fa4092b9993dcf59acdf17 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723a58e548204ab2b3975d7592460e83 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8771d06771b4c3a8278dd35e45d012b : SUCCESS in 22m 33s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/20224247a8ee4235a4f8c8a0abb35407 : SUCCESS in 1h 12m 13s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/158ada66e7bf419585b0cd85e71b879c : SUCCESS in 1h 34m 34s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/cc754fad9d40417f89a4abdcb3f88373 : SUCCESS in 1h 21m 47s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/585ae2ae065c47948329c0a78c3e6313 : SUCCESS in 1h 37m 48s\n- heat-functional-convg-mysql-lbaasv2-amqp1 https://zuul.opendev.org/t/openstack/build/958c966660014c22b56d0cb5105c2bc5 : FAILURE in 1h 30m 37s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/20967385124e4d278279f7f45c3b5ca5 : SUCCESS in 1h 29m 06s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/4636275d406e4edab729f17a6772551a : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":16},{"id":"eb05f18c862f9eba41b5641c62f15f0e274ee001","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-10-08 06:39:40.000000000","message":"Patch Set 16: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"6d84cdf84dfcd22d1f88b55f0da81995610085a6","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-10-08 13:45:06.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1835345c0637d45a779025a3f31479ecda374729","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2019-10-08 16:21:31.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThanks for this patch @Sampat P","accounts_in_message":[],"_revision_number":16},{"id":"7f4caceef098dcefa2f4eeeb5237e6387b4bf9c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 16:21:46.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"b99f4fcc661214148bb2bbd6eb2bcab5ec13b37e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 18:51:00.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86f1549f68cc45aa9e25f0eae1194d70 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72ca430a4c824429b4c6faf7eb80e222 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/179d265e926348868938836c21be492c : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52fd70c985a14ed694fbf7eeb66a4d3e : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88a9111b01eb46b3bfc3afff37eb12cd : SUCCESS in 10m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0090c96b2636457a8a8b2af3d091690d : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/786dd215db3a4219ac72edb217e6e151 : SUCCESS in 14m 10s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/84f7dfc08bb841edb37eb2b4c4f5eb57 : SUCCESS in 1h 22m 43s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/410b14a9d8754b8eaa7ffaaec537c9e6 : SUCCESS in 1h 20m 06s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/cb1f78eb130a4237adb3e69315ed3ce5 : SUCCESS in 1h 33m 50s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/ffab96ebd2304b86b875cc199b8fffef : SUCCESS in 1h 36m 01s\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/0dfe3f2d3c774b8aada87b1cf3efa798 : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":16},{"id":"7c4c2075100b405aeea1a22eb428c9f823853c1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 18:51:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"97cfe885814f6d0634958ea8a22f589b4009da12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 18:53:06.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43e59635fe74389874b40fb54651633 : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4357ae25945745cf85636b96a4fcc484 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":16},{"id":"1915b759bddb77c84f21104a21c2a46354ff45d6","author":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"date":"2019-10-10 13:09:03.000000000","message":"Patch Set 16:\n\n@Zane @Rico, Thanks for the review.","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"169a35e0599a0582dce65c86b0e1a9920ce6bd66","revisions":{"e5bb7b11876248fdc72b80366f761b0b86e5aa7c":{"kind":"REWORK","_number":1,"created":"2019-06-05 19:07:38.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/1"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"subject":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id","message":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\nStory: 2005523\nTask: 30642\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e5bb7b11876248fdc72b80366f761b0b86e5aa7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e5bb7b11876248fdc72b80366f761b0b86e5aa7c"}]},"branch":"refs/heads/master"},"d28beb040197ba072745969af00d53dc16e8bd78":{"kind":"REWORK","_number":2,"created":"2019-06-05 19:18:00.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/2"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:17:45.000000000","tz":-240},"subject":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id","message":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\nStory: 2005523\nTask: 30642\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d28beb040197ba072745969af00d53dc16e8bd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d28beb040197ba072745969af00d53dc16e8bd78"}]},"branch":"refs/heads/master"},"431f0421176f7c8d9ab36d5c8d6759248a7059eb":{"kind":"REWORK","_number":3,"created":"2019-08-09 20:00:09.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/3"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-08-09 19:59:22.000000000","tz":-240},"subject":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id Update Keystone client to support user@domain in user property","message":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id\nUpdate Keystone client to support user@domain in user property\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\nStory: 2005523\nTask: 30642\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/431f0421176f7c8d9ab36d5c8d6759248a7059eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/431f0421176f7c8d9ab36d5c8d6759248a7059eb"}]},"branch":"refs/heads/master"},"ae7d36641453aa7b4ece80ae5ed1a3f567f58bb8":{"kind":"REWORK","_number":4,"created":"2019-08-12 14:42:15.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/4"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-08-12 14:41:20.000000000","tz":-240},"subject":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id Update Keystone client to support user@domain in user property Fixed pep issue","message":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id\nUpdate Keystone client to support user@domain in user property\nFixed pep issue\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\nStory: 2005523\nTask: 30642\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ae7d36641453aa7b4ece80ae5ed1a3f567f58bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ae7d36641453aa7b4ece80ae5ed1a3f567f58bb8"}]},"branch":"refs/heads/master"},"735d0f2c438cdc418d8b5b50dd0fb06ede84b17f":{"kind":"REWORK","_number":5,"created":"2019-08-12 17:12:56.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/5"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-08-12 17:11:30.000000000","tz":-240},"subject":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id Update Keystone client to support user@domain in user property Fixed pep issue Removed new properties user_domain and group_domain","message":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id\nUpdate Keystone client to support user@domain in user property\nFixed pep issue\nRemoved new properties user_domain and group_domain\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\nStory: 2005523\nTask: 30642\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/735d0f2c438cdc418d8b5b50dd0fb06ede84b17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/735d0f2c438cdc418d8b5b50dd0fb06ede84b17f"}]},"branch":"refs/heads/master"},"f8ad4e54e9496a51233314f97926922cec47c53a":{"kind":"REWORK","_number":6,"created":"2019-08-13 15:23:10.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/6"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-08-13 15:22:27.000000000","tz":-240},"subject":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id Update Keystone client to support user@domain in user property Fixed pep issue Removed new properties user_domain and group_domain Updated test case file","message":"Update UserRoleAssignment and GroupRoleAssignment to support domain_id\nUpdate Keystone client to support user@domain in user property\nFixed pep issue\nRemoved new properties user_domain and group_domain\nUpdated test case file\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\nStory: 2005523\nTask: 30642\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f8ad4e54e9496a51233314f97926922cec47c53a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f8ad4e54e9496a51233314f97926922cec47c53a"}]},"branch":"refs/heads/master"},"50c5c0e7d4fdd670346c6d278471f00f1852eb34":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-08-13 15:30:20.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/7"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-08-13 15:26:17.000000000","tz":-240},"subject":"Support domain in user and group search","message":"Support domain in user and group search\n\nImplements: User and group search across domains.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/50c5c0e7d4fdd670346c6d278471f00f1852eb34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/50c5c0e7d4fdd670346c6d278471f00f1852eb34"}]},"branch":"refs/heads/master"},"c81cada494a0464680238575d095996e8b1ee7d8":{"kind":"REWORK","_number":8,"created":"2019-08-13 18:24:46.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/8"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-08-13 18:23:26.000000000","tz":-240},"subject":"Support domain in user and group search","message":"Support domain in user and group search\n\nImplements: User and group search across domains.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c81cada494a0464680238575d095996e8b1ee7d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c81cada494a0464680238575d095996e8b1ee7d8"}]},"branch":"refs/heads/master"},"b370b77e1e0e9d9c8a234e192495b78b88bf0a8f":{"kind":"REWORK","_number":9,"created":"2019-09-04 15:21:11.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/9"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-09-04 15:14:30.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user@domain, group@domain,\nrole@domain and project@domain to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b370b77e1e0e9d9c8a234e192495b78b88bf0a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b370b77e1e0e9d9c8a234e192495b78b88bf0a8f"}]},"branch":"refs/heads/master"},"d80591ad4906df9413509629c0f71f73c173f66c":{"kind":"REWORK","_number":10,"created":"2019-09-05 18:23:15.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/10"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-09-05 18:22:47.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user@domain, group@domain,\nrole@domain and project@domain to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d80591ad4906df9413509629c0f71f73c173f66c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d80591ad4906df9413509629c0f71f73c173f66c"}]},"branch":"refs/heads/master"},"5f742a6ac6c18ab8cd7119cc6a828753998f1984":{"kind":"REWORK","_number":11,"created":"2019-09-09 18:53:33.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/11"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-09-09 18:52:56.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user@domain, group@domain,\nrole@domain and project@domain to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5f742a6ac6c18ab8cd7119cc6a828753998f1984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5f742a6ac6c18ab8cd7119cc6a828753998f1984"}]},"branch":"refs/heads/master"},"ca9089a91760f770ebbe123cf1da8edfb438d66d":{"kind":"REWORK","_number":12,"created":"2019-09-19 18:34:11.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/12"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-09-19 18:33:07.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user{domain}, group{domain},\nrole{domain} and project{domain} to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ca9089a91760f770ebbe123cf1da8edfb438d66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ca9089a91760f770ebbe123cf1da8edfb438d66d"}]},"branch":"refs/heads/master"},"d9e42be9481fbe0a19ece501f231d190965de0f9":{"kind":"REWORK","_number":13,"created":"2019-09-23 14:17:26.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/13"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-09-23 14:17:04.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user{domain}, group{domain},\nrole{domain} and project{domain} to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d9e42be9481fbe0a19ece501f231d190965de0f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d9e42be9481fbe0a19ece501f231d190965de0f9"}]},"branch":"refs/heads/master"},"4135ee10c7e8f50f0950bdc0153a3afcb85dc6fb":{"kind":"REWORK","_number":14,"created":"2019-09-23 14:20:43.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/14"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-09-23 14:20:23.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user{domain}, group{domain},\nrole{domain} and project{domain} to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/4135ee10c7e8f50f0950bdc0153a3afcb85dc6fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/4135ee10c7e8f50f0950bdc0153a3afcb85dc6fb"}]},"branch":"refs/heads/master"},"5d0b9f757ad52b4a4aea3755beaa7f6b7cd17c7b":{"kind":"REWORK","_number":15,"created":"2019-10-01 13:38:58.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/15"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-10-01 13:37:25.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user{domain}, group{domain},\nrole{domain} and project{domain} to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nRelease note added.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5d0b9f757ad52b4a4aea3755beaa7f6b7cd17c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5d0b9f757ad52b4a4aea3755beaa7f6b7cd17c7b"}]},"branch":"refs/heads/master"},"169a35e0599a0582dce65c86b0e1a9920ce6bd66":{"kind":"REWORK","_number":16,"created":"2019-10-01 13:42:53.000000000","uploader":{"_account_id":30515,"name":"Sampat Ponnaganti","email":"sp810x@att.com","username":"sp810x@att.com"},"ref":"refs/changes/04/663404/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/04/663404/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/04/663404/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/04/663404/16"}}},"commit":{"parents":[{"commit":"c4076e12c8dec33873dd47c7faecd76b1d6dd513","subject":"Merge \"Ignore Not Found when deleting Keystone role assignment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c4076e12c8dec33873dd47c7faecd76b1d6dd513"}]}],"author":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-06-05 19:07:09.000000000","tz":-240},"committer":{"name":"Sampat P","email":"sp810x@att.com","date":"2019-10-01 13:42:40.000000000","tz":-240},"subject":"Support domain in keystone lookups","message":"Support domain in keystone lookups\n\nImplements: User, group, role and project lookup across domains.\nAdded domain parameter to keystone lookup functions.\nHeat templates now support user{domain}, group{domain},\nrole{domain} and project{domain} to support cross domain\nlookup. Keystone constrains will also work across domain.\n\nRelease note added.\n\nStory: 2005523\nTask: 30642\n\nChange-Id: I2b02787bd8883ced631b81174cee9134445bf170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/169a35e0599a0582dce65c86b0e1a9920ce6bd66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/169a35e0599a0582dce65c86b0e1a9920ce6bd66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
