)]}'
{"id":"openstack%2Fheat~740937","triplet_id":"openstack%2Fheat~stable%2Fstein~I2a78b6f6235529a95838307e3d38731d31ff5c4b","project":"openstack/heat","branch":"stable/stein","hashtags":[],"change_id":"I2a78b6f6235529a95838307e3d38731d31ff5c4b","subject":"Check external resources after creation","status":"ABANDONED","created":"2020-07-14 10:53:57.000000000","updated":"2022-01-17 19:02:21.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"efec6cf19fef03e788d9650b404e8d51a59e4563","_number":740937,"virtual_id_number":740937,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-16 14:54:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 09:53:31.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2020-09-11 14:12:27.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-09-23 07:41:05.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-11-16 06:34:21.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-11-16 14:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f14ef8fafbb6dd12aff4ce01d18e4c5d10c7e61f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-14 10:53:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"8a11ab84e84fb2fdbdcebc4984feeda122575c8f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-14 10:54:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08fb2635be0dc6d673a5b1a8815bb33f60fd0505","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-07-14 10:54:25.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 3327b40fde0f071b4820c59dd88c6c5565d79406","accounts_in_message":[],"_revision_number":2},{"id":"001052f6f6a6b896cec02472d02de7745f86cf1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 13:09:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d7dd6f3c0314f31bccf5dc7562de66f : SUCCESS in 11m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fd04f31fd914af8801639266f7087ae : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19e11bf98ba4f1a9c600bbca3b1d063 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6694591f7ffa4250a33cce3bdeafda25 : SUCCESS in 10m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9ec42315fbe54037b2e32adc92224c46 : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52f812d78a5e45bd9b602ac1c6bcd678 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0432e23d53cd4b2597bfff4d9d731889 : SUCCESS in 10m 14s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d16309dd34124351bd1b70baa3320126 : SUCCESS in 1h 13m 37s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1bf94a01144847c6b02d9d6541350565 : SUCCESS in 1h 22m 21s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/11b6cf380e2140128428a26d4d9c790c : TIMED_OUT in 2h 12m 01s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/5d22f24ab24c4c33b0161d6cb77160f4 : SUCCESS in 1h 12m 26s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/8c2ef11c2b8b4d1480431545f2e593b9 : FAILURE in 2h 12m 32s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/4e71ba8ea9b64a4ea22a39b503a9dcdb : FAILURE in 2h 10m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e749a7a745cb32d4cff754606b38a3538f96744d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-09-04 15:42:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4fd08dd43b5c8ac5566c96ec5c8dd1b49999c09c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 17:38:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/083a95fcc52d4af4abc6daaafe4367af : SUCCESS in 13m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1192f43328924fa29b407929e7dff5eb : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43072725d29c4493819587ab951f9c7d : SUCCESS in 8m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7676d5354aa4e0d9f819d133c10f85d : SUCCESS in 10m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e114a179c6344d16b3122fa28210019d : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e0eed3bcabb41b990ab431deb88cb2f : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9febb25a074246b0a0690908cf9bd1fc : SUCCESS in 7m 53s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/e9eb8349b69a4cb8a856b7ac57f5147a : SUCCESS in 1h 03m 23s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6db9429de1494185892bd12aa4f52560 : SUCCESS in 1h 49m 52s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/ce4ff7e37e1b43eaa9990df0b9500fc7 : FAILURE in 34m 25s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/9a4767644616459c9cf326b53e1dd5f2 : FAILURE in 37m 19s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/7be9831dd7b64cce9e2a72bea7a47926 : FAILURE in 32m 31s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/b8d18678d84b45a58a78934cebc5bcb0 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ba0987213a742e6255e5d640989fb1e02eb7f65d","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-09-11 09:53:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"380b08080779bf7e84f3b1fad2c9043629bb0865","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-09-11 09:53:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c1f34a17a070e5497ed8165e026762d8cd3db06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 11:24:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62db85a1d7114b22acce062c500522b3 : SUCCESS in 10m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9a1856ec99d4db282f77658f05650d8 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2afb26b35cdc4186a3d9d9dae265f406 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c2e5da7c4cff47dc9347c73d44ed521c : SUCCESS in 7m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8cd5acc6576c45c3a61b57c340cce858 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aefeb66aa3634566a05b5a2d1965f108 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce39431174b84485a85079867addf8cc : SUCCESS in 7m 29s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/118291cd410946feb4b231566bbc23c5 : SUCCESS in 1h 12m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c9c2b216c47a47989f152a8f8c03b326 : SUCCESS in 1h 21m 58s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/93671b4c04374d938366f2ff149750b5 : FAILURE in 34m 19s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/918f0a724381407781ffa391ac836514 : FAILURE in 34m 42s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/0b6d2c2a35a0406c93983297d4462b62 : FAILURE in 29m 37s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/b76ee49dafd84accacf087f0de9f3498 : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":2},{"id":"985ffe2e0decda44b365008acde62cd73b47c575","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-09-11 14:12:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"84d494ce7b7f0b5edc2e3ce657049cbb220f17e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:51:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62743a7611d34f02b1cc58c82e9471c6 : SUCCESS in 9m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/572da1bf616047d18ba8c17b30b6ae2d : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808368a99e5e4576b5085dfe16d02fcd : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6eb2dfd6c1b646739a9f2247bdb3c643 : SUCCESS in 9m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e32600bc1c534565bda1de2f433dfae8 : SUCCESS in 10m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40de7b0da81c47da8cf70842cdd5faf9 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba9586211184aeaa1f068ac7e5fe912 : SUCCESS in 7m 51s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/88b06ab3af9e4e94b6ba456ec8ca4fda : SUCCESS in 1h 06m 58s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/14b876fcee484e83a9b683feb5d4cb44 : SUCCESS in 1h 29m 06s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/8a826c9f887f492b99dd7f888ad4eb90 : FAILURE in 29m 18s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/d91d0648e2c54d8aae81cdcc173b4150 : FAILURE in 31m 44s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/12ec0d5a60614bb9b6fffb2d15c7ed2b : FAILURE in 32m 48s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/0c638f96cc234a95b4ec4ac2cec63f64 : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":2},{"id":"bb495756bc1a023a4ead15bcaeb606a44feb007f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-09-11 18:45:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ad0679f52a949ca222aeff6be91c791d12ac7e72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:11:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3846f6fc4b1e47d0a12b0530b09b486b : SUCCESS in 10m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2cafbad696545b19366af7d145fc9c6 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e99320005bf8458cad3751d4134f280c : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91c1bc34991f4c5da9bf6661d08d5dc5 : SUCCESS in 8m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9038e54691c841c997c5ecb4933b3a1f : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29f12ec65cec4252a68a414f6f78dd0e : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16a12fb987f4b6a9ac53e74acbb6620 : SUCCESS in 7m 05s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/afae4a28ea5f483199bf2f7323baa622 : SUCCESS in 1h 15m 54s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6b72322636d0465c9773afe501e4af87 : SUCCESS in 1h 01m 01s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/1bc18546133947cba7a778df39e21b40 : FAILURE in 30m 52s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/7a3901dd23e74f7487b35e7dbc514790 : FAILURE in 28m 56s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/456a0c2508354ae492f902f2f3922232 : FAILURE in 28m 30s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/b39717be0f8544538bfa8c2ee4cee3e4 : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":2},{"id":"da2aa51c49983bcdde297ee5b9b246c6bfba1a7e","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-09-23 07:40:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fc88c3c15e0794c6af7b6012737852b26c6f7264","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-09-23 07:41:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d5f3ef3240e2303b5ff1d532a190ab016f9a0c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 09:23:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ccea43ddbcd4b4eb3b86c37ad15ac17 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c87975ccc01d412f8e1d0a1dd5cc2cd7 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9def8735a8c943dd98ba6a81d0e7e63d : SUCCESS in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/916228575de44a508c73099b018caae6 : SUCCESS in 10m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0c30b88da6d9483186bcd094d64c6534 : SUCCESS in 9m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edb4a9e6a07e470eb4a791e8ab08fc66 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723bc376271b462a9b4c52186ae3d81c : SUCCESS in 8m 42s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/e4b36c897ebe47e78e37c081191f1299 : SUCCESS in 1h 06m 44s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6b60c83ab669431a9a6be9dffbd42cbb : SUCCESS in 1h 27m 26s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/7c0a2c0e95c540548c38a47478ac2ed8 : FAILURE in 26m 31s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/4e0de6d27e8f4c9c8244a6ae5d009c44 : FAILURE in 25m 45s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/925b299b77df41628ffde79faa2e4a28 : FAILURE in 31m 24s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/e60889bbe1f2434d82bee1e4bb8ab16c : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":3},{"id":"1e8bd6002e7496724ad4fce062385e37bc63fc87","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-09-23 13:52:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7c40afdf57cc9421b96aa9a3eb01d49f226838a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 15:25:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56dfdbf489324301b7c9b8099bfcaacf : SUCCESS in 9m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12816b0b77e54c48a991138bd67e999d : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad20de6edb25470da503f303b2186d23 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/183d00eeb0ad4a85bbd9c2d01bfb4fc2 : SUCCESS in 6m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b7c00afdf5d54f05af06da75a816f3aa : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/753f339cc39e46e48e51cff064b8b9fb : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba95f675808d43f992ee40630a87689c : SUCCESS in 8m 07s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/7d7a430ad5d048a89b348c7e9ef4a099 : SUCCESS in 1h 14m 48s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/bfc1c6a9ab774eed8ce2903a7138c8c7 : SUCCESS in 1h 10m 45s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/de2c5843b9f94e41943479206a05a892 : FAILURE in 28m 22s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/fd1e705f824d4bd3aa5e9ca418c93d46 : FAILURE in 29m 09s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/5bea001f5ab046528dda3e50853445da : FAILURE in 35m 00s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/01d9ca1327094758957c3e92a726cda0 : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":3},{"id":"06b903618715d0f6d60eb826e50288ec698b1521","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-09-30 14:47:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"81c8a926cd230a5730859e9ebf0145c225dd996e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 16:30:04.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d116abb81019488e8dfa8df3948338b3 : SUCCESS in 9m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d217ec4092a48a28a4f867ddfd83040 : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acbbf03ddf0b4b77b719d5b3f9a0c2b7 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9522dff3cc04b7699f5032d4026e1f0 : SUCCESS in 8m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cf2f639394e84c938d3c2b2329fd22cf : SUCCESS in 10m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3228d6733dd042dfb1d93f2ce2704ac9 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e755c98c03a4d2d83cb04c540a19d94 : SUCCESS in 7m 19s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/7aa589fa26a743c58d32241ff686f9c0 : SUCCESS in 1h 02m 27s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0037979665b64b9297abff800ea9e2d1 : FAILURE in 1h 25m 57s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/1d46d145cdae4da9a0f47942f1989de5 : FAILURE in 39m 36s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/37a58850cc154380a449107fcc019169 : FAILURE in 39m 32s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/a6f4e1f6fc2242238657753b647b04d1 : FAILURE in 22m 45s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/5b5f0a7478b34a10afb8e45d80d7bd63 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":3},{"id":"d4cbd3f3b39993b1687090949cebe70b2a73c422","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-10-09 09:00:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8be36bd594f6178a8ef7d10528ff8476fd68f2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 10:32:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29a49e18bdd74638a06df3376b05aefc : SUCCESS in 8m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22b95c87054e47d1b2841d3c1fbab099 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/615dc667c5af4605ae905131c1210811 : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b79a074a47ab4cab87c9f879fb204db3 : SUCCESS in 7m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e93447333e0e4af0af0c9e9c6e183461 : FAILURE in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a524d7a19fb4d8d96842ec95ebdbd6c : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2024d611f1304d2fb31974b732824ed9 : SUCCESS in 8m 05s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/1c1498a2c002411eb664d102fd070f90 : SUCCESS in 1h 10m 20s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/f96351a0f1b7401b84646b4f8d6a6374 : SUCCESS in 1h 19m 02s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/3f50c2de03d4499f878f397a4ace292a : FAILURE in 29m 49s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/0c63f8f7278043e28bc1927362ebf424 : FAILURE in 31m 02s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/0cb4e0ddf00e47b5ba1deffdeae83d12 : FAILURE in 30m 52s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/236ad5c4a06549c7bfb9c1cd6e7d392c : FAILURE in 33m 40s","accounts_in_message":[],"_revision_number":3},{"id":"da3c481419c9e0e84cb469f65fa2b61f9b98bdc5","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-10-23 14:37:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b8970e79bde7cf44ad01bed18d92c8038df4b0f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 16:19:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816d71fffca2475798dd394e69bb2846 : SUCCESS in 10m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e39436edee44cbe8aa75b61da558a66 : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c3fd4204044d30b1e84975c94895d4 : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c82dc9ee3bd545209342593b32f9d93d : SUCCESS in 9m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a3520ea9d0f4a9b9f8cb8a29cd32111 : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a2e1d7b2ad24a27a0c4de9c661db9e4 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67e347274a5647b0a98ea622dfda5b46 : SUCCESS in 8m 31s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/4189fbb8483c45ab811425a8040d36bf : SUCCESS in 1h 09m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/470fddc7ce524b54af672de1df496c6d : SUCCESS in 1h 30m 56s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/7db2d2e314d04ff28e2e10d9ee3f1d2e : FAILURE in 32m 25s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/dd76afbdfda748af9279bfdd58f8c701 : FAILURE in 38m 07s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/4d86b0e74e5f4972974b1cfaefd02d44 : FAILURE in 30m 37s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/e7e1d660c3964d2b8b44e33307f1f138 : FAILURE in 28m 22s","accounts_in_message":[],"_revision_number":3},{"id":"55f6643857d9533e225c83d83d453564ca3694c2","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-10-28 12:40:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d081dfa71076044dca643f90007118c6260095cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-28 16:02:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6855abef2a774c0fa04c40ca32d72c58 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7f1da028a11494fb833f188d7f5cb77 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd1857580ed4da296806cf8b6f4de96 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a52343dee034697b24366fafb75f1be : SUCCESS in 12m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/66666c6d9b614db48a6c9ad2ea663d8e : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd0111b890234b92a8bfe7149b5ce0ff : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4cbc6eb3b774aa3bbf494f1157a81d8 : SUCCESS in 6m 29s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/09b7e1f61d7d4f129914a6db118a5cac : SUCCESS in 1h 12m 32s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/62fd5a3b7ba44b6fb34da3e7783e975b : SUCCESS in 1h 33m 39s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/4010cc333d8e408c88c116e90f4b2acb : FAILURE in 51m 07s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/d7cc8eab030240caa54f4003b993e34b : FAILURE in 48m 06s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/1f7f58f4c237413daee1bc67dfc27351 : FAILURE in 48m 22s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/3be2bff45dd2497f9003505aa7eca102 : FAILURE in 35m 05s","accounts_in_message":[],"_revision_number":3},{"id":"40b212311a24fc9a8be61b08b31a9be8548ecb6c","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-11-16 06:34:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5cd1e2d803f093f63abd316075fb9b528e92dbd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 08:05:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e04170178d5b4a47921250817468d0f4 : SUCCESS in 8m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b18ee083aa74752afcde33b73ee013e : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07c13e9cc93452ab2867ecb06c7f7f6 : SUCCESS in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bb953e2584845a9907b4f1be403dd1e : SUCCESS in 11m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6187eee7104f4613842bc7291b6af130 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f878af5687e44fcd80bebf5b28f3dc94 : SUCCESS in 9m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/379e3b0db5d0492baeb2fbe94dad0b14 : SUCCESS in 7m 04s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/0780efa27ee24a358e62d3bc55316266 : FAILURE in 39m 36s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4dba70929a4147809d75eb37e92e3016 : FAILURE in 56m 41s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/b740ca9ba82e4c56b2aaf7f8ab4a795e : FAILURE in 43m 04s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/01d48069361f42ffbac0cfc84fefc6f1 : FAILURE in 41m 36s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/d413930693bc4db1b5241dff42f6aabf : FAILURE in 44m 53s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/148cbf83e69f4e9ba2c49153de27909d : FAILURE in 30m 07s","accounts_in_message":[],"_revision_number":3},{"id":"d6b77dc1290520c28811648025b8654ddda36c43","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-11-16 13:12:49.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a06bae3e374ccbbc8dada34ea306be966b740411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 14:54:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0b3a39ee8a947d28b57957bfe042f4b : SUCCESS in 9m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/059e8f8d6cb0472dba52c75fe316c566 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e15245c7e9248c788b175ab2344af47 : SUCCESS in 8m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d44f41b7887f497395bb89b0125ebe0d : SUCCESS in 7m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0ac1a4b4fbbd4df6839938b172e0423e : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ad1f0be7b6448dad9b34db3164e40a : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eba9abbe9a34d63a41a475b02a7fed5 : SUCCESS in 7m 25s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/0f971abee180482dadf9a4be3d1a6882 : FAILURE in 48m 22s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/68a0f9b8974e4f118f721815e81ca8aa : SUCCESS in 1h 25m 07s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/746acbd87fea464587ee6acb34d47832 : FAILURE in 34m 40s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/5caf26198d5c4dad9c05cdcf9b5cf23e : FAILURE in 33m 08s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/80826dc8e1cd4693a7a03453e690a447 : FAILURE in 36m 04s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/0b8fa09f4db2442b834e116f030e2595 : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":4},{"id":"d64512f7f58b58da08f747336d9d3a0a0336fb98","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-01-17 14:31:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"efec6cf19fef03e788d9650b404e8d51a59e4563","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2022-01-17 19:02:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9bf11e37f52e87611353c268ccf32342cbfba06f","revisions":{"06a467635b0c55e9e9bd1dc1440ee93f31af5e35":{"kind":"REWORK","_number":1,"created":"2020-07-14 10:53:57.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/37/740937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/37/740937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/37/740937/1"}}},"commit":{"parents":[{"commit":"8f1bcbaf840297430a95637449598e7de405289e","subject":"Fix Tempest installtion on system wide for stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/8f1bcbaf840297430a95637449598e7de405289e"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-18 04:04:44.000000000","tz":-300},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2020-07-14 10:53:57.000000000","tz":0},"subject":"Check external resources after creation","message":"Check external resources after creation\n\nThe method Resource.check() is not a generator function, but it does\nreturn a generator in most cases. Simply calling the function does not\ncause the resource to actually be checked. Only in the case where the\nresource did not provide a handle_check() method would it do anything at\nall - and in that case it simply changes the state to CHECK_COMPLETE.\n\nYield the generator function so that the resource will actually be\nchecked if the resource type supports it.\n\nChange-Id: I2a78b6f6235529a95838307e3d38731d31ff5c4b\nTask: 37828\n(cherry picked from commit 5c6038f7a272238eef8978a4b26ab4ae4efd745b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/06a467635b0c55e9e9bd1dc1440ee93f31af5e35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/06a467635b0c55e9e9bd1dc1440ee93f31af5e35"}]},"branch":"refs/heads/stable/stein"},"3fdfc34d22744a602ee61029c3eb68f10f74a369":{"kind":"NO_CHANGE","_number":2,"created":"2020-07-14 10:54:10.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/37/740937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/37/740937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/37/740937/2"}}},"commit":{"parents":[{"commit":"8f1bcbaf840297430a95637449598e7de405289e","subject":"Fix Tempest installtion on system wide for stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/8f1bcbaf840297430a95637449598e7de405289e"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-18 04:04:44.000000000","tz":-300},"committer":{"name":"Kevin Carter (cloudnull)","email":"kevin@cloudnull.com","date":"2020-07-14 10:54:10.000000000","tz":0},"subject":"Check external resources after creation","message":"Check external resources after creation\n\nThe method Resource.check() is not a generator function, but it does\nreturn a generator in most cases. Simply calling the function does not\ncause the resource to actually be checked. Only in the case where the\nresource did not provide a handle_check() method would it do anything at\nall - and in that case it simply changes the state to CHECK_COMPLETE.\n\nYield the generator function so that the resource will actually be\nchecked if the resource type supports it.\n\nChange-Id: I2a78b6f6235529a95838307e3d38731d31ff5c4b\nTask: 37828\n(cherry picked from commit 5c6038f7a272238eef8978a4b26ab4ae4efd745b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/3fdfc34d22744a602ee61029c3eb68f10f74a369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/3fdfc34d22744a602ee61029c3eb68f10f74a369"}]},"branch":"refs/heads/stable/stein"},"db91e9647fa3273c79200272bc313740f8c4abb5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-23 07:40:56.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/37/740937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/37/740937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/37/740937/3"}}},"commit":{"parents":[{"commit":"65a93b121e2161591b679f688c1f6dfa43d39c3d","subject":"Merge \"Use never expiring alarm_url for resource attributes\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/65a93b121e2161591b679f688c1f6dfa43d39c3d"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-18 04:04:44.000000000","tz":-300},"committer":{"name":"Adriano Petrich","email":"apetrich@redhat.com","date":"2020-09-23 07:40:56.000000000","tz":0},"subject":"Check external resources after creation","message":"Check external resources after creation\n\nThe method Resource.check() is not a generator function, but it does\nreturn a generator in most cases. Simply calling the function does not\ncause the resource to actually be checked. Only in the case where the\nresource did not provide a handle_check() method would it do anything at\nall - and in that case it simply changes the state to CHECK_COMPLETE.\n\nYield the generator function so that the resource will actually be\nchecked if the resource type supports it.\n\nChange-Id: I2a78b6f6235529a95838307e3d38731d31ff5c4b\nTask: 37828\n(cherry picked from commit 5c6038f7a272238eef8978a4b26ab4ae4efd745b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/db91e9647fa3273c79200272bc313740f8c4abb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/db91e9647fa3273c79200272bc313740f8c4abb5"}]},"branch":"refs/heads/stable/stein"},"9bf11e37f52e87611353c268ccf32342cbfba06f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-11-16 13:12:49.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/37/740937/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/37/740937/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/37/740937/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/37/740937/4"}}},"commit":{"parents":[{"commit":"65a93b121e2161591b679f688c1f6dfa43d39c3d","subject":"Merge \"Use never expiring alarm_url for resource attributes\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/65a93b121e2161591b679f688c1f6dfa43d39c3d"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2019-12-18 04:04:44.000000000","tz":-300},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-11-16 13:12:46.000000000","tz":0},"subject":"Check external resources after creation","message":"Check external resources after creation\n\nThe method Resource.check() is not a generator function, but it does\nreturn a generator in most cases. Simply calling the function does not\ncause the resource to actually be checked. Only in the case where the\nresource did not provide a handle_check() method would it do anything at\nall - and in that case it simply changes the state to CHECK_COMPLETE.\n\nYield the generator function so that the resource will actually be\nchecked if the resource type supports it.\n\nDepends-On: https://review.opendev.org/#/c/762834\n\nChange-Id: I2a78b6f6235529a95838307e3d38731d31ff5c4b\nTask: 37828\n(cherry picked from commit 5c6038f7a272238eef8978a4b26ab4ae4efd745b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9bf11e37f52e87611353c268ccf32342cbfba06f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9bf11e37f52e87611353c268ccf32342cbfba06f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
