)]}'
{"id":"openstack%2Fheat~745902","triplet_id":"openstack%2Fheat~stable%2Fussuri~I0137bd73549ca47529a1707f9a3d077fc679ff8d","project":"openstack/heat","branch":"stable/ussuri","hashtags":[],"change_id":"I0137bd73549ca47529a1707f9a3d077fc679ff8d","subject":"[DNM] Test Heat CI","status":"ABANDONED","created":"2020-08-12 14:14:55.000000000","updated":"2020-08-20 18:16:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3ebd3ed81cd23089f7c1a982c197f13c6dc7c51b","_number":745902,"virtual_id_number":745902,"owner":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-12 16:08:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"all":[{"value":-1,"date":"2020-08-12 14:15:56.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 16:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9dc903c46d45c197d2cc5b43412ce78fde3ee0f","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-12 14:14:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"761cea975c8e78be50d69bdbbf8a623f7c2f2a67","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-12 14:15:56.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b7a1736e6db1ffd4e7b9d64ca4ec85ed76a4c760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 16:08:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b32d91ed424242eead06e000f4946126 : SUCCESS in 14m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a75e7e087874b27aec96141ba51ddfd : SUCCESS in 13m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c91b0cdc7db4f9fba4b1292bf033171 : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa79a1740d704100b7214c8888148a05 : SUCCESS in 10m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af92fb073d46465e8cbbd07d1ec4a28f : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9f545e104446acbc33a50cd544a0cf : SUCCESS in 11m 26s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/6614fede28c34942ac04b6cab31ee2e2 : SUCCESS in 1h 37m 11s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1e8ca27d0dd54a15a7d3c067f9e621fd : SUCCESS in 1h 45m 24s\n- heat-functional https://zuul.opendev.org/t/openstack/build/6247e190dad04660ac47239bfafb577c : SUCCESS in 1h 02m 24s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/13ede26a560d4788888bf7f7633f0eab : SUCCESS in 1h 02m 40s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/48512a39e1d5471b8c71fc6b48112bcc : SUCCESS in 1h 09m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ebd3ed81cd23089f7c1a982c197f13c6dc7c51b","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-08-20 18:16:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9962a14c17e5dd862ff10155bdd4b32c6033fe87","revisions":{"9962a14c17e5dd862ff10155bdd4b32c6033fe87":{"kind":"REWORK","_number":1,"created":"2020-08-12 14:14:55.000000000","uploader":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"ref":"refs/changes/02/745902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/02/745902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/02/745902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/02/745902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/02/745902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/02/745902/1"}}},"commit":{"parents":[{"commit":"ec6cd9ed001f46198684c7b477a4e2334236d474","subject":"Merge \"Check external resources after creation\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ec6cd9ed001f46198684c7b477a4e2334236d474"}]}],"author":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-12 14:14:02.000000000","tz":-240},"committer":{"name":"Luke Short","email":"ekultails@gmail.com","date":"2020-08-12 14:14:41.000000000","tz":-240},"subject":"[DNM] Test Heat CI","message":"[DNM] Test Heat CI\n\nChange-Id: I0137bd73549ca47529a1707f9a3d077fc679ff8d\n(cherry picked from commit 17d4fce1652d1713362bf6473d37429dc6c9b050)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9962a14c17e5dd862ff10155bdd4b32c6033fe87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9962a14c17e5dd862ff10155bdd4b32c6033fe87"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
