)]}'
{"id":"openstack%2Fheat~753335","triplet_id":"openstack%2Fheat~stable%2Fvictoria~Ie987f60222c1690fe3b180385f8ee891adea0fb1","project":"openstack/heat","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"Ie987f60222c1690fe3b180385f8ee891adea0fb1","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","status":"MERGED","created":"2020-09-22 14:11:43.000000000","updated":"2020-11-02 11:11:38.000000000","submitted":"2020-11-02 11:05:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753335-1604315114402-c7592bc5","meta_rev_id":"5785de1f15a5616eb55c4a717405d5d8ef4a3b8a","_number":753335,"virtual_id_number":753335,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-10-26 13:45:15.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-02 11:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-22 14:35:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2020-09-27 03:48:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2020-10-19 11:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":1,"date":"2020-10-19 11:43:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-22 14:35:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-27 03:48:10.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2020-10-19 11:43:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-10-26 13:45:15.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-11-02 11:05:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae32765af21c83fc12f5e6fca6b52ea266d9592a","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-22 14:11:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea39cd37f315beddd382aec89f3f4eb225d9cb7a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-22 14:35:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd857a3d65cc5021596d802ed45c017be050e8d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 14:53:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd08e712d20442da9f8fbcd2652f8502 : SUCCESS in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a611c020582437e9f795995cbaeb6cb : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904d3f5964fb407b8af80ea964877f16 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c617a8c77ef14e38b57b5df014f38253 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ef311fface9492590add7a50a59c3ec : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a48c8ddc2624f0890a3411a68673b4f : SUCCESS in 18m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66b0646ad6dd4f7882c039ffff2dd017 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0edbcd500c7dd0fd2b41b1e1784db09811d88e62","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-09-27 03:48:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6231a8ecf5a2993b139cc97a7597f9f38c700fb6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-19 11:43:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3714a091ea4162941a4faa2246fd873c841abdb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 11:44:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a9d31ce819508b1e09b6956262c4f712c26c794","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 13:33:49.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b28bda461f380e8ce7dadd2e47370e7b2f142823","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-10-26 08:28:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5bc39cf952b83c94e56102e4bfca7e96929e0064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 10:12:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19c0c193aeda4a51a7064565fb6af827 : SUCCESS in 8m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fbb5b4681534f72a524f489f9925efc : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e32049f63d648a4920c0747eaf03614 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/038fca1e0d3743a68a40c371c6b82244 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60f80ae77d32491bb6bf1bc330f17bce : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344da644ad294976a63e09672088754a : SUCCESS in 10m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68d79c70278b42f08c2911cc99f27ea4 : SUCCESS in 9m 08s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/22ac1ebb4d334cee9e2755e116a9fbaa : FAILURE in 38m 10s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9e6eb18dd960434b8fb1d22c820170d5 : SUCCESS in 1h 29m 42s\n- heat-functional https://zuul.opendev.org/t/openstack/build/829cda4ba8cf464d8cc0053385ee560d : SUCCESS in 1h 04m 58s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/6b18368ef01b485a8a2911ce4d35269a : SUCCESS in 58m 21s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/4a05538747104ab4a7a3f832979be3ec : SUCCESS in 1h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"074733a5f6e865eb1cb1cb439f96e706c8560d0c","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-10-26 13:45:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2decbc7084327ff6c2a0cc08a8c15ebca452a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 16:17:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f33b7edc4d2e462cb4974dca0617b428 : SUCCESS in 37m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a869d7b6cb94888bff1d1eea9ad070b : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d98b8719dd74740a808843b95733edd : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd6a9234f8a046e786de2f28fbc271b5 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0009e21e1860431ea37d6a4606ca3395 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/388c8d4ee2c34dfdbdb26d0be4afbe6c : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57ca288d32014a6d922c7397150a5b99 : SUCCESS in 24m 32s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/20e1f6cd44a14966a78cfe85057a3640 : SUCCESS in 1h 14m 03s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c176575ddca64090ba862a854a29da87 : SUCCESS in 1h 33m 00s\n- heat-functional https://zuul.opendev.org/t/openstack/build/0f8e8d6e54d6483891a0fa5e3ab49ba9 : SUCCESS in 1h 07m 30s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/f455cc1ce3824e6794133f3e093f6573 : TIMED_OUT in 2h 12m 53s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/ab5549583dbe4f90bece46b6e4c7f02a : SUCCESS in 1h 02m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b494d595e738bf9918a35ded565100b1070937c9","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-10-30 04:17:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"377fe3ebb23b93c2efcb72e694a694c89e5d7fc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-30 05:52:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f0ce47fa2fd49bda7b84be0a40e082d : SUCCESS in 13m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2fff46b4a4746b992bd492c2188a19f : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa0fe5aa82a49a5947777f2a49d713d : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfe5beb1979c44b2be667b87478a7a73 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7014eaf9988d48b7a172cfe3bebc0273 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7ba36240b8b421eaf3b9956bab236a8 : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/474939fee6154577a4a6c8523597bc9f : SUCCESS in 10m 57s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/225cf0acf44949d8be253d5afbcc3ee2 : SUCCESS in 1h 24m 25s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/ee175f00f121444a9df793750774daa8 : SUCCESS in 1h 31m 00s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b30f9ae94eb64afd88cfb9bae1b0e868 : FAILURE in 1h 14m 05s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/3c13dec5c3ca40a8af53fce3675c5175 : SUCCESS in 57m 06s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/76e3b53c7f794ecd844d86e364415c2a : FAILURE in 55m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bffc22c46d47731088799477b32002655cd690f6","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-11-02 07:58:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4309c1087d8414f470139202e9d3399be806941a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 09:30:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac2cdd3a02a1427898f8994e997fbdb5 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df30baa6ea46484693a9cbbc9a7fdce7 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443aaab368b14e908a2c15c9cfb02f7d : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91e5f30219cb4b19a23a01682297b71e : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/124f169a1cc24c85aa5d81d10a64a53f : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/539e2ae94d3b4a99b384a52e168fb1f3 : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c26c1768eea4b2e91b28168c085d4a6 : SUCCESS in 8m 00s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/f52de5656bc84734bd513d8aec2e7182 : SUCCESS in 1h 16m 54s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/41129d23281247e9a7ee2c89ea0fa68f : SUCCESS in 1h 27m 08s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c934b43b836e49c98579961178f79794 : SUCCESS in 1h 03m 35s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9b8a12cffbcc46bfa4fb4ee98bcf3f08 : SUCCESS in 1h 02m 38s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/fc633c2a72a242fe943c733bf6134053 : SUCCESS in 1h 07m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be6726599dca7e3070aac9b7d16e0545d942f2ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 09:31:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9762be1c880c2d696050da8808af371b14d87f46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 11:05:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b63b5f0fa424592aa2818c2c464291c : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390adb93c7884cd4986774c5cb99232b : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cbb9db866d24d219e2819281575899c : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48d1affb6cbe41ddaa280a76a769a5a5 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d333342fec1d454ea29eae5d184392a1 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0773dba47704409684716cbfa9ed1ce0 : SUCCESS in 7m 47s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/e619e3a61c9f43eb8cef1c0dc9e12824 : SUCCESS in 1h 26m 20s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0f2b7c7b495f45eeb9ab95b9a80d46fe : SUCCESS in 1h 25m 03s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d2e5686963ea48cd9e5c6625891e8e03 : SUCCESS in 58m 09s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/db5b728ef60e4e8b95ad094be66e7222 : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a2d211c0debca4f01875193d6467717a4c5cb653","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 11:05:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd7f0279391ae4bdf999ebfdcbb35e9630e47325","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 11:11:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d31d545f973470886f4a83960994f92 : SUCCESS in 1m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25dcf9480cee4d35a5a7134862c56e05 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f1ab43bb8420cdf72231ea0f996ed04796354e0","revisions":{"5f1ab43bb8420cdf72231ea0f996ed04796354e0":{"kind":"REWORK","_number":1,"created":"2020-09-22 14:11:43.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/35/753335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/35/753335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/753335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/753335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/753335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/35/753335/1"}}},"commit":{"parents":[{"commit":"dc6602079d357067b686031b04ff8c868252ba35","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/dc6602079d357067b686031b04ff8c868252ba35"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-22 14:11:41.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-22 14:11:41.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","message":"Update TOX_CONSTRAINTS_FILE for stable/victoria\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/victoria branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ie987f60222c1690fe3b180385f8ee891adea0fb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5f1ab43bb8420cdf72231ea0f996ed04796354e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5f1ab43bb8420cdf72231ea0f996ed04796354e0"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
