)]}'
{"id":"openstack%2Fheat~761203","triplet_id":"openstack%2Fheat~stable%2Fvictoria~I627f2df7f6135b3d194e7146ae45c3b7f8b6394d","project":"openstack/heat","branch":"stable/victoria","hashtags":[],"change_id":"I627f2df7f6135b3d194e7146ae45c3b7f8b6394d","subject":"Configure swap as before","status":"MERGED","created":"2020-11-03 17:13:27.000000000","updated":"2020-11-04 18:32:14.000000000","submitted":"2020-11-04 18:19:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761203-1604513940728-aa5a3b26","meta_rev_id":"d12daa648d8ab9d0d92a80f2766a231e17eb41ac","_number":761203,"virtual_id_number":761203,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"date":"2020-11-03 17:13:56.000000000","_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-04 18:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2020-11-04 16:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"date":"2020-11-03 17:13:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2020-11-04 16:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 17:13:27.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2020-11-04 16:39:22.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-11-04 18:19:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d54e646ae6761d061bb72cfc0212032a90c5619e","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-11-03 17:13:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0f8eea67138cea240213e5a9ef2eb617955e9e66","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2020-11-03 17:13:56.000000000","message":"Patch Set 1:\n\noriginal patch missed victoria branch, ussuri cherry-pick is already merged","accounts_in_message":[],"_revision_number":1},{"id":"06002bf98ba51708bec9ce98df756c583d8ba3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 18:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4356df4bf8d24d88934824530f6f85e0 : SUCCESS in 8m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f95c62bac81348f8b889bdddc5140237 : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cfb67bf741d403b8862d1ac6857b83b : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2cfe37eb85845c680f211f28c375584 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da67fb418559422790deed41334af3c3 : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252fee9a1b524b569e97ec6358cbd443 : SUCCESS in 13m 08s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/090db1e1f404402ea96aed0187828de4 : SUCCESS in 1h 24m 45s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/fd19539a6a8b4cc5a81715891a65deaf : SUCCESS in 1h 33m 52s\n- heat-functional https://zuul.opendev.org/t/openstack/build/de1ba902cc4f4b9dbec3b0c84f3b07ca : SUCCESS in 1h 03m 04s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/4a43221c44184819b07afeb878218f5c : SUCCESS in 1h 25m 41s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/7025332fc945427389420408b7e07a67 : SUCCESS in 59m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3e81a00ee4a42218adf4ea1d1ed7cdaaf2616f5","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-11-04 16:39:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4a123443d3767d1c70204912caf500a03fb5aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 16:39:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac786b1d105bcd32ac008facf4f0cd732e39a41e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 18:19:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/614435e913b64ea8869e1d9c1b66da0a : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc835107a084b1098e619df9fd09d7c : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bfec68cd06141e887a6b8bab487f38b : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/573201a15d0449df9e40de131b925597 : SUCCESS in 19m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452e6d014c2b4427a42de3777c0e42e1 : SUCCESS in 11m 02s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/864f48f359dc47cba2edaa0b4dad7c7d : SUCCESS in 1h 14m 15s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/5463e0de4c5d4243b8e7d63eda37854e : SUCCESS in 1h 31m 30s\n- heat-functional https://zuul.opendev.org/t/openstack/build/6ae4adbcc4b54d9d95b1e3051de36ee3 : SUCCESS in 1h 04m 13s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/8bb5fc8b5b634213a373a36b1acfedc4 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a18c20b69c0a1f3a154c7a0767f215fb5f013a55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 18:19:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d12daa648d8ab9d0d92a80f2766a231e17eb41ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 18:32:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169871e03ae04404bac5170a86546cae : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e30ea0af5e96a501a5a3a3948e40efac0e7c087d","revisions":{"e30ea0af5e96a501a5a3a3948e40efac0e7c087d":{"kind":"REWORK","_number":1,"created":"2020-11-03 17:13:27.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/03/761203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/03/761203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/03/761203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/03/761203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/03/761203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/03/761203/1"}}},"commit":{"parents":[{"commit":"5f1ab43bb8420cdf72231ea0f996ed04796354e0","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5f1ab43bb8420cdf72231ea0f996ed04796354e0"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-10-09 07:09:10.000000000","tz":330},"committer":{"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","date":"2020-11-03 17:13:27.000000000","tz":0},"subject":"Configure swap as before","message":"Configure swap as before\n\nhttps://review.opendev.org/750941 changed the default\nswap to 1024MB for jobs and this seems to resulting in\nhigh failure rate of the functional and grenade jobs.\n\nLet\u0027s configure it back to what used to before i.e 8192MB.\n\nChange-Id: I627f2df7f6135b3d194e7146ae45c3b7f8b6394d\n(cherry picked from commit f419923a0d3aa62485296ab7666485e74af57c8e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e30ea0af5e96a501a5a3a3948e40efac0e7c087d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e30ea0af5e96a501a5a3a3948e40efac0e7c087d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
