)]}'
{"id":"openstack%2Fheat~762952","triplet_id":"openstack%2Fheat~stable%2Fqueens~If7cb4fbc83f5aa62a0b73e003ff23ec2b40b2ded","project":"openstack/heat","branch":"stable/queens","topic":"bug/1692445-stable/queens","hashtags":[],"change_id":"If7cb4fbc83f5aa62a0b73e003ff23ec2b40b2ded","subject":"Sync support network type for provider network","status":"ABANDONED","created":"2020-11-17 05:26:34.000000000","updated":"2021-11-24 03:46:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01391186c524e2af6820adaf3e92dc4ac20d8467","_number":762952,"virtual_id_number":762952,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-17 06:32:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-17 06:32:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd95ef2bff1c79e8e8dc80798a19a40357944c8e","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-11-17 05:26:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3b72c81216bb5aad21c37f007e3fa769c2ab270b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 06:32:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa80e259c3484188846c10914b3cd398 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16c6da5728664dad8f01fdf0e1abf3da : SUCCESS in 7m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ee122adc711e4b28ab3ce72abf9f91f0 : SUCCESS in 6m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c64b4cf1439a4da7af91ed7e36270cb6 : SUCCESS in 7m 01s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ebddd3adbff941658957327b714eb729 : FAILURE in 54m 04s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/966dd9c9a78c4a2caa585c814e46471c : SUCCESS in 53m 52s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/a0ed04b631734c69bf47ca82649e7b5c : FAILURE in 38m 36s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/27b21f4f9a6a4b208e3b1c28a096afea : FAILURE in 40m 05s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/cba753c2b28b4bb9b50022616551c125 : FAILURE in 35m 22s\n- heat-functional-convg-mysql-lbaasv2-py35 https://zuul.opendev.org/t/openstack/build/ede7507cbf024b43bbcd722284ec34b5 : FAILURE in 28m 40s","accounts_in_message":[],"_revision_number":1},{"id":"01391186c524e2af6820adaf3e92dc4ac20d8467","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-11-24 03:46:39.000000000","message":"Abandoned\n\ntagging eol","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f41d0add0b4d7a426cacb288b405fc172f69e7a8","revisions":{"f41d0add0b4d7a426cacb288b405fc172f69e7a8":{"kind":"REWORK","_number":1,"created":"2020-11-17 05:26:34.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/52/762952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/52/762952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/52/762952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/52/762952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/52/762952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/52/762952/1"}}},"commit":{"parents":[{"commit":"14da305bf1a67ea1914a0b44d3ce3ee14f19399f","subject":"Merge \"Allow scale-down of ASG as part of update\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/14da305bf1a67ea1914a0b44d3ce3ee14f19399f"}]}],"author":{"name":"huangtianhua","email":"huangtianhua@huawei.com","date":"2017-05-22 08:33:14.000000000","tz":480},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-11-17 05:26:34.000000000","tz":0},"subject":"Sync support network type for provider network","message":"Sync support network type for provider network\n\nNow heat only supports \u0027vlan\u0027 and \u0027flat\u0027 network type\nfor OS::Neutron::ProviderNet, this will add all other\nsupported network types: \u0027vxlan\u0027, \u0027local\u0027, \u0027gre\u0027 and\n\u0027geneve\u0027.\n\nChange-Id: If7cb4fbc83f5aa62a0b73e003ff23ec2b40b2ded\nCloses-Bug: #1692445\n(cherry picked from commit 4e6bb6febc4ff730fa82cca329c9218c708f7ecc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f41d0add0b4d7a426cacb288b405fc172f69e7a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f41d0add0b4d7a426cacb288b405fc172f69e7a8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
