)]}'
{"id":"openstack%2Fheat~762964","triplet_id":"openstack%2Fheat~stable%2Ftrain~Ief22c7038eef2b7f9c6070e166917cfa612d262a","project":"openstack/heat","branch":"stable/train","topic":"check-py27-health","hashtags":[],"change_id":"Ief22c7038eef2b7f9c6070e166917cfa612d262a","subject":"[TEST][DMT] Checking py27 under train branch","status":"ABANDONED","created":"2020-11-17 07:16:31.000000000","updated":"2020-11-18 15:27:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b67ec5c938d75c669956e44c0d13f6600361c6e4","_number":762964,"virtual_id_number":762964,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-11-18 01:40:26.000000000","_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-18 03:03:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-18 03:03:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4406cac4b422995fb1fe4603c9eb62e81f9d5234","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-11-17 07:16:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85826a71e0291e68d7734d8a29196377e95e4d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:23:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/e0a099bf28b348e28dfa55d8a73cdfb5 : FAILURE in 2h 04m 41s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/89cedd97a2124a7a8e5a48620cbe88c4 : SUCCESS in 1h 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c8a4eae17ff128bccb150905f204141c787a1c0c","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-11-18 01:40:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2d0b9fb318e2920d690385f7cf0618b80e2811c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 03:03:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/0ae9970ca1e148dfbe6e25f0f79f73d4 : SUCCESS in 1h 19m 02s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/fffedcf9959c46e0b82b76afbc0ff126 : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6cc3f0fe3c171c6bad08a6ae16c7b2e5dea04aff","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-11-18 15:27:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"769446b4fa3bacec6cd9ec47e70b64d63a86311d","revisions":{"769446b4fa3bacec6cd9ec47e70b64d63a86311d":{"kind":"REWORK","_number":1,"created":"2020-11-17 07:16:31.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/64/762964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/64/762964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/64/762964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/64/762964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/64/762964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/64/762964/1"}}},"commit":{"parents":[{"commit":"0d7e5bde1eae6c17d9a6697eca9c813c6de2c5a4","subject":"Merge \"Fix OS::Heat::Delay resource type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0d7e5bde1eae6c17d9a6697eca9c813c6de2c5a4"}]}],"author":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-11-17 07:15:45.000000000","tz":480},"committer":{"name":"ricolin","email":"rico.lin.guanyu@gmail.com","date":"2020-11-17 07:15:45.000000000","tz":480},"subject":"[TEST][DMT] Checking py27 under train branch","message":"[TEST][DMT] Checking py27 under train branch\n\nChange-Id: Ief22c7038eef2b7f9c6070e166917cfa612d262a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/769446b4fa3bacec6cd9ec47e70b64d63a86311d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/769446b4fa3bacec6cd9ec47e70b64d63a86311d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
