)]}'
{"id":"openstack%2Fheat~765751","triplet_id":"openstack%2Fheat~stable%2Fussuri~Id869f83a46454897c2fe7a532eebfa2863befe5e","project":"openstack/heat","branch":"stable/ussuri","topic":"victoria-f32-stable/ussuri","hashtags":[],"change_id":"Id869f83a46454897c2fe7a532eebfa2863befe5e","subject":"Replace Fedora test image with F32","status":"MERGED","created":"2020-12-08 04:14:14.000000000","updated":"2021-01-19 07:47:44.000000000","submitted":"2021-01-19 07:42:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"765751-victoria-f32-stable/ussuri","cherry_pick_of_change":765539,"cherry_pick_of_patch_set":2,"meta_rev_id":"29227800f8b957212ed67fdb550aa2168da5c80a","_number":765751,"virtual_id_number":765751,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 07:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-01-19 05:57:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-01-18 16:47:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-01-19 05:57:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 04:14:14.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-12-08 05:09:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 05:47:51.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2020-12-22 15:11:41.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2021-01-14 19:55:01.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"0aa7400f49e70bc1583ca7f6985a108e489abe71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-12-08 04:14:14.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* devstack/lib/heat","accounts_in_message":[],"_revision_number":1},{"id":"5a4692dba6b948891cf5730a78ae4b4c92e1358f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 05:09:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5171605c40064a159a7e50f69c8d7041 : SUCCESS in 9m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36b857b7a5424f999fc0592c1a2ab5af : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38663ecfedab4a3898d601e82f2bbc65 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7553fb6bcd8410585a3b50647b8276c : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22fd3495b0734c1786b8a9fce4e52402 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90499173ce314c88a196097ef1dde93e : SUCCESS in 11m 46s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/c53fe559409b4ff0bf85d96a12b0dfce : FAILURE in 53m 00s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4dee8388c18a495d8ec5cf8b641a71be : FAILURE in 52m 23s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/413c8965307f470fa79bd22edb88b3a4 : FAILURE in 15m 29s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/834f3251012046fb913806a780061804 : FAILURE in 18m 45s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/b0a69edc146c47f6915eb2821ba9de4c : FAILURE in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdcaaac7b87b578b803a1d6e15c4e784e85dfc5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-12-08 05:46:48.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aee4818659bb302d6aae82f9674e0c3d42955e8f","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-12-08 05:47:51.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"2f55a2d35571a3ddb0a21187dd3fa81fd88d2d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 07:27:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94d9fea4a8374d05ab44ebef3ea3c6b5 : SUCCESS in 13m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/baa144c7d11342b99cb4aada5bcb7440 : FAILURE in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/406f21725ece425a960549443ac0b149 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2738a923e97b4d55838448c645a145b3 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3ea642eaac742298210560f86a4993b : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049c85f99d2d4dc595442c032d321559 : SUCCESS in 21m 05s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/6b4a8676ad4a4ade938d38cc754ef9a7 : FAILURE in 1h 22m 27s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9250bbfd3adc400b8605ee4dc762e8e4 : SUCCESS in 1h 29m 55s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/90ddcc1881634ee29e92670b54bba3e0 : SUCCESS in 1h 36m 52s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/11dadc560a974055bc9866b075f73966 : SUCCESS in 1h 30m 52s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/e7747910945547b59d903e4e72b39ade : SUCCESS in 1h 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2098a572c622249bd59ac62f57b6bed01feab470","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-09 07:19:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8221a29ba1b4afa8ba7f4fa305345a88f8b5521f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 09:00:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cfce9110a6742f980659c24c1175e71 : SUCCESS in 12m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e43737138294bfa9030972743fd99bb : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d5490117a64ed09a9fbf76da13cd98 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/113c8cf63cf1418fa72418f7903662cd : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8665d560dffa4e8888bfda5e6decd2e9 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8d564d00c74d4db3c77349a81cee52 : SUCCESS in 10m 34s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/c5fd0243cf794bfc93595dd8df8893a3 : SUCCESS in 1h 20m 19s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/dd54bec83a84493492f36fb71f50f43a : SUCCESS in 1h 37m 39s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/9c6a58e98c2446878afe3df2105f2b22 : SUCCESS in 1h 05m 45s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/77f76ce644914f42984c0346746d0339 : SUCCESS in 1h 03m 36s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/c3705a4228eb43a99a2e0139b15ecd44 : SUCCESS in 1h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b40599a838880c1e92e64784a6cafc64a41f0346","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-12-09 09:31:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nlower-constraints job is also broken.","accounts_in_message":[],"_revision_number":2},{"id":"32ce24a88dacf97448a09e4dc0041c74e7ead89d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-12-10 05:38:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"340be9aea3cef5f201e841df683baaddf2d8472e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 08:39:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/632fde219ec54f46939373b1aef0518d : SUCCESS in 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80873cef09b0435ab2eccad7d6beda1c : FAILURE in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae913a6fdff4aaeb5adc3eb1f82477c : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/830b2132468a44e9990993751f4f9c23 : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10bb18dc01ca4b36be4f2904fafde257 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0130c849000c4a15bd762e7be54044ce : SUCCESS in 10m 20s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/9d91204f13184e698f433ae43e709530 : SUCCESS in 1h 31m 44s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/173e6d922a5942cbb6ae0d0fe1818b04 : SUCCESS in 1h 47m 37s\n- heat-functional https://zuul.opendev.org/t/openstack/build/ae02ce611ca246c5b37223e60176f5b6 : SUCCESS in 1h 05m 43s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d3b174c6c6cb49f785c26805441321c7 : SUCCESS in 1h 01m 44s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/849ebfb70bf2456ab56752e5c1c1909b : ERROR Failed to update project None in 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ceb162a0b4d176765d4d087ff4aed057ab3ed3d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-12-22 15:11:41.000000000","message":"Patch Set 3: Code-Review+1\n\nCan someone look at this vitrage gate is failing because of this patch not merged","accounts_in_message":[],"_revision_number":3},{"id":"3817c4ef734ee8e74a5f81cf3f982852f7d79e94","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-01-12 07:45:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"aa29ac31ad2430a4999efd6249b4ec58d1fa9145","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 09:27:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/128d501d5aba424284643fed6c1f1351 : SUCCESS in 9m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a259deb17e5f4c5daa776d558f0b28b9 : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca9de794c024e9c92894495f1444250 : FAILURE in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a28a1cf80e243b2bdf7ef3486dea817 : FAILURE in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/63d2cd2edfe84f2a98b8e52b605692ed : FAILURE in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32f4c9ef6914a828ebb44859942dade : FAILURE in 7m 01s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/e8e01bc374ce43408a8b168e409e276d : FAILURE in 6m 01s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/06cc15de78b0442bb4a192b63af0b273 : SUCCESS in 1h 32m 51s\n- heat-functional https://zuul.opendev.org/t/openstack/build/ae5d10e639db4109b2f9bb612dd81372 : FAILURE in 5m 34s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/44137410efc24dd68d45741439c9d591 : FAILURE in 6m 00s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/9e33fc742c2e48979d2fe1b5cb47ff9b : FAILURE in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6b871ef8647035876fb5453dfae17d9061b38c5c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2021-01-14 10:42:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"deff3c6c24379d94432d0f784b1a5035bbfa162b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 12:58:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3409ea7ed60e48939c03c57293169392 : SUCCESS in 15m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbc5b06aeeeb4e01b624460c6a80b318 : FAILURE in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcefb1d7dac544f68e040e991059121f : FAILURE in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6110b68c2fe84a00b73ffe3b0d9729db : SUCCESS in 8m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e41797cb7b5e47179d305470de680078 : SUCCESS in 21m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37687902e864a019fdb2d4a95642430 : SUCCESS in 11m 41s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/184c96fa023143c989bd6d65ad0d46c0 : SUCCESS in 1h 21m 04s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/08f633953ebf4f308d69a7eb78d408a9 : SUCCESS in 2h 00m 06s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c468722775bd42d2a4c530cf689265f5 : SUCCESS in 1h 38m 26s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/af0608d0ca2846ccb61043c958a4f4e6 : SUCCESS in 1h 05m 01s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/70636d934eb1437abc2f59ed5bb2329b : SUCCESS in 57m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"942c73cdf9e479c17201633a2e9445b5ce9f6495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-14 19:55:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1fa377ad7c429d3e73619ccb9d18c83158262cf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-14 20:08:46.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c473ef85bd334128450c5e25cb5027573f029158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 21:51:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59eb3500e0bc40bca17116ba26bbd537 : SUCCESS in 14m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54caac1710e5424babf38bbdce428cd4 : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b04c1f10c3c24fcd85a822e68e964591 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90166497503543039b0938fce399f226 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d83060ae2c543e590abc565b6ed6d17 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f26619cdde9347e3a50efb856c67f893 : SUCCESS in 11m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9aa8cebddf194553830cc90f3686a59d : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d0caf1d24774123adedf58d15685f0b : SUCCESS in 6m 46s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/5f17aafebf91425eba954964cebc3088 : FAILURE in 1h 28m 19s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6cf20d9f6a2b4aa9ae7ae9dfbccf252c : SUCCESS in 1h 40m 01s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c85e164266564244b944a7c60bcc9861 : FAILURE in 1h 36m 57s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7addfb338571484387b9eab56a0023e8 : SUCCESS in 1h 05m 42s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/4934d63457ef4cfb844b9b37ab72247b : SUCCESS in 1h 02m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6b0e3f6d123d0bdbca1b1e6a8e888e03f7aceb20","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-15 06:39:18.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"9e74877060875c3b77a645f3c9db934b51a0a95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 08:48:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/611f328f0e924b58a9e06aca18cd79e3 : SUCCESS in 11m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ff1157f2ec44e0ab8c90380551c766b : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f35ee6ec7b4232913f0e5c5dc46787 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe5752d0c88f4d5ea24b50c87925996d : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4007873306e849ae967e32313d2b8ec7 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae2a438d8e264b9d9028591e5ede80a3 : SUCCESS in 12m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b0bbc777df5d46049fa98043d13b4e08 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/123595742fde474f88e4192cbc8ce1a0 : SUCCESS in 8m 06s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/4ef47793c078492cbf091697e61506f9 : SUCCESS in 1h 28m 27s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/36da27ba59fd4fd7b6c1d81631516ea5 : SUCCESS in 1h 35m 52s\n- heat-functional https://zuul.opendev.org/t/openstack/build/cd2b555f531e43418dba84ec407a4e2c : SUCCESS in 1h 04m 03s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/f76af3d7aa6c41a0bd7ec80335dae8a1 : SUCCESS in 1h 06m 56s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/1d307d5e651e4d26885c8d2484ee3ef4 : SUCCESS in 57m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eb8693c21ad69a33df2c3c44c33957da07b0668e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-18 16:47:02.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8c08c9ae582616308105e9c95981140e3847966b","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-01-19 05:57:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d19a705fa333c8abe3d89cab54ff7b27f0d22f96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 05:58:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3f4b90a7ed2fd4fd89b2065bd156bc1c475dd908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 07:42:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dbe65409c304bb1952fba1682f730ea : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cbff1aaeeec424a86dcb79452b095ab : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/886f9faf62b34a9390db30d1aaaf973c : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/752e32285a6a410587551cfe76b0f007 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4549ed951a546a89ec4b0efbc6811f4 : SUCCESS in 12m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0bdf557a59da4752b6e8c8c493c495e0 : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c4fff3db4224ba6bcbc44fd263e9641 : SUCCESS in 8m 29s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/148d9353effb4aa486f3ff782a6b2c62 : SUCCESS in 1h 32m 02s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/67ec303255224282b3e21d8d8d327f14 : SUCCESS in 1h 38m 42s\n- heat-functional https://zuul.opendev.org/t/openstack/build/11d978018b7b41639c16f309095279ad : SUCCESS in 1h 02m 54s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/1477314198624930aba79397439c1a77 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":5},{"id":"9913f847ca73c2cedf3da5cf216c938f16349472","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 07:42:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"29227800f8b957212ed67fdb550aa2168da5c80a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 07:47:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875d78e86c0f417ea758fdfd118b1871 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e405aff4e8e4c64a85b8a486d9503f6 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e16fbbd13f02ae2f902e077f5d01560773fdfab6","revisions":{"47db4a596adea5f4853046dace4bd0af0f5070ea":{"kind":"REWORK","_number":1,"created":"2020-12-08 04:14:14.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/51/765751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/51/765751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/51/765751/1"}}},"commit":{"parents":[{"commit":"ec67396a2ba9e9bca14b26f1daf88ae55eabe05f","subject":"As of nova microversion 2.57, personality is deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ec67396a2ba9e9bca14b26f1daf88ae55eabe05f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-12-03 13:07:05.000000000","tz":60},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-12-08 04:14:14.000000000","tz":0},"subject":"Replace Fedora test image with F32","message":"Replace Fedora test image with F32\n\nFedora 31 was retired and the image is gone from mirrors.\nheat-cfntools have been dropped from fedora images, disable\nthe test till that\u0027s resolved.\n\nAlso makes grenade job non-voting, till this is backported\nto stable/victoria.\n\nChange-Id: Id869f83a46454897c2fe7a532eebfa2863befe5e\n(cherry picked from commit 61e53c4eca3bc4f0c60754269cb2811afc569705)\n(cherry picked from commit 779b39c975c32e8fd0a3982197787c294cf92589)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/47db4a596adea5f4853046dace4bd0af0f5070ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/47db4a596adea5f4853046dace4bd0af0f5070ea"}]},"branch":"refs/heads/stable/ussuri"},"57f837ec51fafc396ae98bdd2a6f71927d3b5eca":{"kind":"REWORK","_number":2,"created":"2020-12-08 05:46:48.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/51/765751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/51/765751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/51/765751/2"}}},"commit":{"parents":[{"commit":"ec67396a2ba9e9bca14b26f1daf88ae55eabe05f","subject":"As of nova microversion 2.57, personality is deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ec67396a2ba9e9bca14b26f1daf88ae55eabe05f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-12-03 13:07:05.000000000","tz":60},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-12-08 05:46:36.000000000","tz":0},"subject":"Replace Fedora test image with F32","message":"Replace Fedora test image with F32\n\nFedora 31 was retired and the image is gone from mirrors.\nheat-cfntools have been dropped from fedora images, disable\nthe test till that\u0027s resolved.\n\nAlso makes grenade job non-voting, till this is backported\nto stable/victoria.\n\nChange-Id: Id869f83a46454897c2fe7a532eebfa2863befe5e\n(cherry picked from commit 61e53c4eca3bc4f0c60754269cb2811afc569705)\n(cherry picked from commit 779b39c975c32e8fd0a3982197787c294cf92589)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/57f837ec51fafc396ae98bdd2a6f71927d3b5eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/57f837ec51fafc396ae98bdd2a6f71927d3b5eca"}]},"branch":"refs/heads/stable/ussuri"},"b7f118613d39e2fb546f829f179eec66b38b645d":{"kind":"REWORK","_number":3,"created":"2020-12-10 05:38:27.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/51/765751/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/51/765751/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/51/765751/3"}}},"commit":{"parents":[{"commit":"ec67396a2ba9e9bca14b26f1daf88ae55eabe05f","subject":"As of nova microversion 2.57, personality is deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ec67396a2ba9e9bca14b26f1daf88ae55eabe05f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-12-03 13:07:05.000000000","tz":60},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-12-10 05:26:12.000000000","tz":0},"subject":"Replace Fedora test image with F32","message":"Replace Fedora test image with F32\n\nFedora 31 was retired and the image is gone from mirrors.\nheat-cfntools have been dropped from fedora images, disable\nthe test till that\u0027s resolved.\n\nAlso makes grenade job non-voting, till this is backported\nto stable/victoria.\n\nChange-Id: Id869f83a46454897c2fe7a532eebfa2863befe5e\n(cherry picked from commit 61e53c4eca3bc4f0c60754269cb2811afc569705)\n(cherry picked from commit 779b39c975c32e8fd0a3982197787c294cf92589)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b7f118613d39e2fb546f829f179eec66b38b645d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b7f118613d39e2fb546f829f179eec66b38b645d"}]},"branch":"refs/heads/stable/ussuri"},"a0b9b865e040c70b7218bb47b6fe6b0ea2f473e2":{"kind":"REWORK","_number":4,"created":"2021-01-14 19:55:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/51/765751/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/51/765751/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/51/765751/4"}}},"commit":{"parents":[{"commit":"ec67396a2ba9e9bca14b26f1daf88ae55eabe05f","subject":"As of nova microversion 2.57, personality is deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ec67396a2ba9e9bca14b26f1daf88ae55eabe05f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-12-03 13:07:05.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-14 19:53:15.000000000","tz":60},"subject":"Replace Fedora test image with F32","message":"Replace Fedora test image with F32\n\nFedora 31 was retired and the image is gone from mirrors.\nheat-cfntools have been dropped from fedora images, disable\nthe test till that\u0027s resolved.\n\nbandit removed B322 test, so it needs to be removed from tox.ini, too.\n\nFix contradicting lower constraints to make lower-constraints job pass.\n\nChange-Id: Id869f83a46454897c2fe7a532eebfa2863befe5e\n(cherry picked from commit 61e53c4eca3bc4f0c60754269cb2811afc569705)\n(cherry picked from commit 779b39c975c32e8fd0a3982197787c294cf92589)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a0b9b865e040c70b7218bb47b6fe6b0ea2f473e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a0b9b865e040c70b7218bb47b6fe6b0ea2f473e2"}]},"branch":"refs/heads/stable/ussuri"},"e16fbbd13f02ae2f902e077f5d01560773fdfab6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-01-14 20:08:46.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/51/765751/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/51/765751/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/51/765751/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/51/765751/5"}}},"commit":{"parents":[{"commit":"ec67396a2ba9e9bca14b26f1daf88ae55eabe05f","subject":"As of nova microversion 2.57, personality is deprecated","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/ec67396a2ba9e9bca14b26f1daf88ae55eabe05f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-12-03 13:07:05.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-14 20:06:20.000000000","tz":60},"subject":"Replace Fedora test image with F32","message":"Replace Fedora test image with F32\n\nFedora 31 was retired and the image is gone from mirrors.\nheat-cfntools have been dropped from fedora images, disable\nthe test till that\u0027s resolved.\n\nNOTE(elod.illes): the following changes were made compared to the\noriginal patch:\n\ngrenade jobs don\u0027t need to be non-voting anymore.\n\nbandit removed B322 test, so it needs to be removed from tox.ini, too.\n\nFix contradicting lower constraints to make lower-constraints job pass.\n\nChange-Id: Id869f83a46454897c2fe7a532eebfa2863befe5e\n(cherry picked from commit 61e53c4eca3bc4f0c60754269cb2811afc569705)\n(cherry picked from commit 779b39c975c32e8fd0a3982197787c294cf92589)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e16fbbd13f02ae2f902e077f5d01560773fdfab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e16fbbd13f02ae2f902e077f5d01560773fdfab6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
