)]}'
{"id":"openstack%2Fheat~766178","triplet_id":"openstack%2Fheat~stable%2Fvictoria~Ie524c54e3b982bc6b0786c875d34d177444ec6fc","project":"openstack/heat","branch":"stable/victoria","hashtags":[],"change_id":"Ie524c54e3b982bc6b0786c875d34d177444ec6fc","subject":"Align lower-constraints for new pip","status":"MERGED","created":"2020-12-09 10:47:37.000000000","updated":"2020-12-10 09:42:37.000000000","submitted":"2020-12-10 09:41:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766178","cherry_pick_of_change":766124,"cherry_pick_of_patch_set":3,"meta_rev_id":"7b6e277f1219a1010d61d5c537d63180de7fb55b","_number":766178,"virtual_id_number":766178,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-10 09:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2020-12-10 05:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2020-12-10 05:38:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2020-12-10 05:40:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 11:56:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 05:38:06.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"}],"messages":[{"id":"b36ea5bfd1e1f90f4d6d089b6dd0b99c6267ecb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-12-09 10:47:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"858a7910589192b55efd526f74d1c6c191985734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 11:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8665c57301b14771bd299629e0b03d50 : SUCCESS in 9m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64b4684d7c79420d800e7b4835c55d4f : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4dd2e30bf640ccabdcbc13dcf0d189 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82ed9980082c4978aaf493f2da7f052b : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1097afb03e3b493b8c687a426fe0ca60 : SUCCESS in 8m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d350015509417ba4ea9ef7a39306d6 : SUCCESS in 11m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7806b9e333a64a20a979f032238aaf6b : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8a791f65d4d4301843d20a4a4f3f5e7 : SUCCESS in 9m 00s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/f22fb803929f4168b4ef02830117c220 : FAILURE in 34m 20s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/3883c9423a794c27bfa9b82580e2e035 : FAILURE in 48m 44s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/3d521c10c896489b8eb16ee398965454 : SUCCESS in 1h 06m 46s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/b187dbba9a6b447285e5f04685ac7f2b : SUCCESS in 58m 19s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/f2689c93e5144b8a93614b3dd7ec594f : SUCCESS in 1h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7ca777eb30f12388cf45079935b1a76141155fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-12-10 05:35:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d46f5f39ded8074df7f7d4999729b0773f81d18","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2020-12-10 05:38:06.000000000","message":"Patch Set 2: Code-Review+1\n\nActually prety weird that we have to update lower-constraint and requirement for stable branch because new pip, but that\u0027s the only way to test old branch with new pip","accounts_in_message":[],"_revision_number":2},{"id":"081f0d2f835d35dfcaf1e5598acc452ab72605b4","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-12-10 05:40:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"027bcd8b961bec1c69024dad8b9adf43beb1b14c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 05:40:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1dff4908f8c0c5583b9af9be170d6b2b982b2bc","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-12-10 05:43:20.000000000","message":"Patch Set 2:\n\n\u003e Actually prety weird that we have to update lower-constraint and requirement for stable branch because new pip, but that\u0027s the only way to test old branch with new pip\n\nAFAIK these are issues with lower-constraints (ignored earlier) which should be fixed in master and stable branches.","accounts_in_message":[],"_revision_number":2},{"id":"a63e34bcc951c0af32a34c53395472a4eeadb05b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 07:58:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2add9fb4cb26454eab3c6286ff9bfb0a : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1dfb91584b42c8aac44b6bb04498be : SUCCESS in 8m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a2c7b3607c741998a0a6002e50d49de : SUCCESS in 9m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b0ea5d23b248278e3d079db64d9f6a : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b9333d618940b5b23cd9d0d425f639 : SUCCESS in 14m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/15aa6f1c6b314049a431269df7ddc8b6 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74aae7a7c4894fd58deabf83a3155c63 : SUCCESS in 10m 18s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/8d7dbdd750c5479e970197c713a0bacc : FAILURE in 51m 50s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/544c8a63e66a4e35bc99ce427ea81f4a : ERROR Failed to update project None in 5s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/9446cb061d054ef5a2bcd3497f9911c8 : SUCCESS in 1h 09m 49s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/967b52d372db483eae7e5104e4a80b8b : ERROR Failed to update project None in 4s","accounts_in_message":[],"_revision_number":2},{"id":"c990e2c8e3c91b283a9ca09604269b530e690e7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 07:58:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d3237c3233a4ec78c1c118d74be69f9 : SUCCESS in 9m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aa0bb86594247ce82d7f7931e7ddb99 : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd47d3a256d642db859d486cb496e5eb : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/181af9cd2953478290476c3d178b429b : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf1ac1c59910476b961dfc9f05cc1708 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/173dbf9c35ce478eae1e2d1f23fd7adc : SUCCESS in 10m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/186b8f9be7f64bcc86a388a94d5c7f51 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39f85a0a7ccc40ecbfda81574c1eadb8 : SUCCESS in 6m 14s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/bfe89ad48a594f4ba8e779a946baa722 : FAILURE in 33m 54s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/46990b8237ae49669a073c0d191419b8 : FAILURE in 39m 28s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/1710846061454ff585d476485466bb56 : SUCCESS in 54m 50s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/b87f3056483d47ae8d18c12b703293cd : SUCCESS in 54m 26s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/1424ab457db8419cbb3a92411b68f93d : SUCCESS in 1h 07m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60ba8377482624bd0fc1e2a28db68c55ff68d630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 07:59:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cdc46bef0b742c5808681bbae2e57eec1b5ecb2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 09:40:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e65a8eb365d34fd483af9ff5f4b0b9d7 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d0f7fa41d64de2a20befc29fa30edf : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/178f4565a5e4463e91431dbe2b7b3424 : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b8d62ff7005497cb1eab00c0dc6e621 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70868af383fb40f4a4e4ec32b438b713 : SUCCESS in 11m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4bf1c815d4664cc798ff4829653c5887 : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03afc47de40c4cce8a5e6b17ac5371c5 : SUCCESS in 13m 17s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ae8036bc031f41d99c1d541f24c8134d : ERROR Failed to update project None in 3s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/02e78a7985f64c0eb4fa99480742eece : FAILURE in 38m 14s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/97d7586dc4864dc4a63d0ccd333da988 : SUCCESS in 1h 34m 37s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ac673922c1ee49b5b02f555d644320d3 : SUCCESS in 1h 30m 49s","accounts_in_message":[],"_revision_number":2},{"id":"282569cf3786bf4a4d476f4ad59c648aace24261","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 09:41:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7b6e277f1219a1010d61d5c537d63180de7fb55b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 09:42:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13a5c07732f4b00b1043932dd084356 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d4c39e1230f42beab310cf68a608ef1 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e03502a6e3477dcada86f51adeff8d22d2e5951","revisions":{"395cc730d1f1fd1278cb274843dc59b4225ad64a":{"kind":"REWORK","_number":1,"created":"2020-12-09 10:47:37.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/78/766178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/78/766178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/78/766178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/78/766178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/78/766178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/78/766178/1"}}},"commit":{"parents":[{"commit":"779b39c975c32e8fd0a3982197787c294cf92589","subject":"Replace Fedora test image with F32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/779b39c975c32e8fd0a3982197787c294cf92589"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-12-09 03:57:16.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-12-09 10:47:37.000000000","tz":0},"subject":"Align lower-constraints for new pip","message":"Align lower-constraints for new pip\n\nNew pip version is quite strict and does not allow conflicting\nminimum version deps in lower-constraints.\n\nChange-Id: Ie524c54e3b982bc6b0786c875d34d177444ec6fc\n(cherry picked from commit 4370af157262d953ed32c857a9ddb06968f3ab58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/395cc730d1f1fd1278cb274843dc59b4225ad64a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/395cc730d1f1fd1278cb274843dc59b4225ad64a"}]},"branch":"refs/heads/stable/victoria"},"7e03502a6e3477dcada86f51adeff8d22d2e5951":{"kind":"NO_CHANGE","_number":2,"created":"2020-12-10 05:35:26.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/78/766178/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/78/766178/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/78/766178/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/78/766178/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/78/766178/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/78/766178/2"}}},"commit":{"parents":[{"commit":"779b39c975c32e8fd0a3982197787c294cf92589","subject":"Replace Fedora test image with F32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/779b39c975c32e8fd0a3982197787c294cf92589"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-12-09 03:57:16.000000000","tz":330},"committer":{"name":"Rico Lin","email":"rico.lin.guanyu@gmail.com","date":"2020-12-10 05:35:26.000000000","tz":0},"subject":"Align lower-constraints for new pip","message":"Align lower-constraints for new pip\n\nNew pip version is quite strict and does not allow conflicting\nminimum version deps in lower-constraints.\n\nChange-Id: Ie524c54e3b982bc6b0786c875d34d177444ec6fc\n(cherry picked from commit 4370af157262d953ed32c857a9ddb06968f3ab58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/7e03502a6e3477dcada86f51adeff8d22d2e5951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/7e03502a6e3477dcada86f51adeff8d22d2e5951"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
