)]}'
{"id":"openstack%2Fheat~769971","triplet_id":"openstack%2Fheat~stable%2Ftrain~I0f8d55e39132a1d357033d460089a9775ad0b515","project":"openstack/heat","branch":"stable/train","topic":"cap-bandit-1.6.2","hashtags":[],"change_id":"I0f8d55e39132a1d357033d460089a9775ad0b515","subject":"[stable-only] Cap bandit and fix lower constraints job","status":"ABANDONED","created":"2021-01-08 17:47:52.000000000","updated":"2021-02-24 16:51:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d3d916710c74986f2b6cc9aa439395cbf8814764","_number":769971,"virtual_id_number":769971,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-08 19:12:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 19:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f274a7489f99e15d6e9872c2590a89a07da64d59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-01-08 17:47:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11c55ce4802d0e707aefe3d880d947e1ab33e6e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 19:12:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f5bbe88ef7a4ed09f94a765d4651961 : SUCCESS in 10m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c5dd946d723425ca5cdaedc1816c855 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c032179eb9b240acbba617e2cfed085f : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/891c0d17bab744e2a9a7630fd66c8b24 : SUCCESS in 9m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a41f5c61f0c3458f9b2d32426ce439b0 : SUCCESS in 8m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba93b8cc308242f7834cdcdf3cba8b2a : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc09a36ba9dd49e0b0d59b8e84704f78 : SUCCESS in 7m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce2712560200476a89875e7a5227ba8c : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5788cfd603b94ae7a6cf5cc5fe9641b2 : SUCCESS in 8m 15s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/f3d270cbf2ac46c3a6ee2805619337ce : SUCCESS in 1h 16m 03s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/da0540a9a2cd441e88d88f2b08096b24 : SUCCESS in 1h 14m 14s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/f2ddeb57bec24478b570439ed5d3b67a : FAILURE in 37m 27s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/95023546739e48b599b9d27c029bb943 : FAILURE in 38m 30s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/905d5458ac124a40b3ecbb809fad6686 : FAILURE in 36m 01s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/223ea0323c674f18b143f90fdf231262 : FAILURE in 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d3d916710c74986f2b6cc9aa439395cbf8814764","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-24 16:51:09.000000000","message":"Abandoned\n\nAbandon as this patch was squashed into https://review.opendev.org/c/openstack/heat/+/771819 and it was merged.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7fa936e12ffc57662c024c70fecaa2d7f389be4","revisions":{"d7fa936e12ffc57662c024c70fecaa2d7f389be4":{"kind":"REWORK","_number":1,"created":"2021-01-08 17:47:52.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/71/769971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/71/769971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/71/769971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/71/769971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/71/769971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/71/769971/1"}}},"commit":{"parents":[{"commit":"0d7e5bde1eae6c17d9a6697eca9c813c6de2c5a4","subject":"Merge \"Fix OS::Heat::Delay resource type\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0d7e5bde1eae6c17d9a6697eca9c813c6de2c5a4"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 17:42:32.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-01-08 17:42:32.000000000","tz":60},"subject":"[stable-only] Cap bandit and fix lower constraints job","message":"[stable-only] Cap bandit and fix lower constraints job\n\nThe 1.6.3 [1] release has dropped support for py2 [2] so cap to 1.6.2\nwhen using py2.\n\ncryptography version needed to be bumped as it conflicts with another\ndependency and causes the new pip resolver to time out at gate.\n\npygments needed to be added to lower-constraints to make\nrequirements-check job pass.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nChange-Id: I0f8d55e39132a1d357033d460089a9775ad0b515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d7fa936e12ffc57662c024c70fecaa2d7f389be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d7fa936e12ffc57662c024c70fecaa2d7f389be4"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
