)]}'
{"id":"openstack%2Fheat~774270","triplet_id":"openstack%2Fheat~stable%2Ftrain~I93abdbf8a9532af1aec7ad15a6f384beb9fc3ff9","project":"openstack/heat","branch":"stable/train","topic":"wait-exp-stable/train","hashtags":[],"change_id":"I93abdbf8a9532af1aec7ad15a6f384beb9fc3ff9","subject":"Use random exponential delay","status":"MERGED","created":"2021-02-06 06:17:00.000000000","updated":"2021-02-08 04:49:11.000000000","submitted":"2021-02-08 04:47:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774270-wait-exp-stable/train","cherry_pick_of_change":694485,"cherry_pick_of_patch_set":3,"meta_rev_id":"a02e143ee181a9c2f62d80df635cee36f70528ce","_number":774270,"virtual_id_number":774270,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-08 04:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-02-08 03:23:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-02-06 10:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-02-08 03:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 06:17:00.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2021-02-06 07:47:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cd0fe715eaed9cf854b1bfa0875031919f52f2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-06 06:17:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e2c48a5f66103e64385b7ebb6c69a5f22761fb1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 07:47:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25a5ba01e32042f3a9ec7d0f38688420 : SUCCESS in 10m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37a8d4d075554b639c6915957c296ecc : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae8f66ce8d004edeb71960953a9323cb : SUCCESS in 11m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1db4738609b647b3b59670d3f655b8d5 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5542aff14c740efb91b514d7224f538 : SUCCESS in 12m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38412dd19d6947a48a0ef3ced980eb9a : SUCCESS in 9m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735ed9d990bc4877a57f1e32289fca26 : SUCCESS in 8m 00s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/9ac51383ebec43d39ef058e496ee820c : SUCCESS in 1h 07m 03s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/5798e867c1d74fbf993677db8e8d71c3 : SUCCESS in 1h 08m 25s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/76d20233912248728a68aea0255637c1 : SUCCESS in 1h 23m 38s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/25c531750ec142abbe581455c19a4a04 : FAILURE in 1h 06m 50s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/784667c25d0245eb8d6f534cc2a35652 : SUCCESS in 1h 08m 57s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/190a568835dc455bbb893bf8196a7b54 : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6a4cf01e82189d4c934a8b4d8601b7e39de29849","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-06 08:42:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13fdfc4588be0575ce6d38238f580a98d6a2b128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 09:40:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ce70777e2ac493faa57111036f9d3e3 : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f554e21c9a146328875330fa9b0621b : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22f4baf9f71b41efa902bff9c957afdd : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac7b14cb49d941739c87404d9171023e : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5aceece25c184f7481557e26bb6e7572 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ce3b0b93e4e4f42ac551a0e2544aaa8 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a359f846d749589d383f5f89ffee7a : SUCCESS in 5m 58s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/49e1dc34ce4c4bdba03b30f580201af4 : SUCCESS in 52m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/8afa16798303405088175fb1e207b57c : SUCCESS in 55m 23s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/49ffbd85374540a588029a9ffa6b1fbc : SUCCESS in 54m 53s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/798669baf6964e1aa6de30e71c0d2042 : SUCCESS in 49m 04s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/7cd48b178a82458181fc683a1c4a569d : SUCCESS in 55m 24s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/99f54c1ef28a463086dfd51886b4f148 : SUCCESS in 45m 28s","accounts_in_message":[],"_revision_number":1},{"id":"74030c474f3bbbcb352addd9db8539728e882f06","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-02-06 10:15:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"828fa73c6d71b04532c10ee160816c87f2322d24","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-02-08 03:23:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nApproving as it\u0027s a minor backport.","accounts_in_message":[],"_revision_number":1},{"id":"31e16367029cdfd905ee8557281c8095cea2afb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 03:23:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8272ca78198bb1d9b8a828a437f272cfb886b292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 04:47:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c256e0acbff343f0817a0aaea061dc17 : SUCCESS in 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4196b60e0982476d8f7bbc2415f7d6bb : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/018ec623ccf141759056ebcf2e3ab64d : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82acf3fb71e9487089fd3f6e133723ec : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f2547fd143a4354a785b080edf187ba : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c66c6c0fe04a25b470622e66175f1f : SUCCESS in 6m 27s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/a6a0a19e815c4456947c17f873927e8a : SUCCESS in 1h 13m 11s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/d496cc13131e4c8bb8120c9c7a0d37ff : SUCCESS in 1h 12m 36s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/632dee36fd82406c8156b4ea38a98c8f : SUCCESS in 1h 17m 22s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/cc654de3bfa54ced92e67415a56976c2 : SUCCESS in 1h 12m 12s\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/fb0e8948b5c84eb496a874d31c4248cd : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":1},{"id":"28c4e11a6bf050583125c36a4c736021997dcfc6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 04:47:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a02e143ee181a9c2f62d80df635cee36f70528ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 04:49:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22cd42b7e0e54d15bbe3c6f30e1b9600 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"069c4db8fed218702c9cd08d773c5ea9a0807fb0","revisions":{"069c4db8fed218702c9cd08d773c5ea9a0807fb0":{"kind":"REWORK","_number":1,"created":"2021-02-06 06:17:00.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/70/774270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/70/774270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/70/774270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/70/774270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/70/774270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/70/774270/1"}}},"commit":{"parents":[{"commit":"fb2e723653bb4ac13322605c1dc5e2661b97c5b9","subject":"Replace Fedora test image with F32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/fb2e723653bb4ac13322605c1dc5e2661b97c5b9"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2019-11-07 12:03:07.000000000","tz":120},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2021-02-06 06:17:00.000000000","tz":0},"subject":"Use random exponential delay","message":"Use random exponential delay\n\nthis is much more suited for resolving contention when\naccessing a (lockable) shared resource.\n\nChange-Id: I93abdbf8a9532af1aec7ad15a6f384beb9fc3ff9\n(cherry picked from commit 56f6e583d19ef4dd0c1462bfece677a6ec2fa756)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/069c4db8fed218702c9cd08d773c5ea9a0807fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/069c4db8fed218702c9cd08d773c5ea9a0807fb0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
