)]}'
{"id":"openstack%2Fheat~780701","triplet_id":"openstack%2Fheat~master~Ifc8c19e181902566d4f295fa979ab6869a4e0852","project":"openstack/heat","branch":"master","topic":"2008718","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-10-04 06:18:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifc8c19e181902566d4f295fa979ab6869a4e0852","subject":"Allow deleting user_creds when can\u0027t be decrypted","status":"MERGED","created":"2021-03-16 05:07:07.000000000","updated":"2022-10-04 06:20:14.000000000","submitted":"2022-10-04 06:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780701-2008718","meta_rev_id":"b46b20e9724a9dcb033cd19cbf5b237b71778728","_number":780701,"virtual_id_number":780701,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-04 06:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-08-30 06:03:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-08-30 11:22:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-10-04 04:48:56.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-10-04 04:48:56.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 05:27:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-16 06:54:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 06:03:13.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-08-30 11:22:44.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-10-04 04:48:56.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"}],"messages":[{"id":"0a354602bc21d62d1c2ef6117f29ae1af9da8c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-03-16 05:07:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e030caab2c6b2c5b09f3eb99fad46582abc045ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 05:27:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e452595d71344d7ea03e35f01ba8feac : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db8c4dba486e4cde9f22034a746ce11b : SUCCESS in 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ef1a7cd64d00cc7044b0129ba429f0d068aaeaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 06:54:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b148b68f88a5469eaf188790a3f5f7f0 : SUCCESS in 11m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed4310d44f534249962ccc14d23d54f9 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23eece8208bf4ae58ac9468c71cee0c3 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab1c09456aa94f1aa7f61bb3b80e6b2e : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/696930ef005c4e05aac2f06d3182a279 : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b505cfb1e5174cf1adb1804c84098856 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d531332d3894fdf9745cf5e104f637d : SUCCESS in 14m 01s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1ea3113af4884a8496502275788e7e5e : FAILURE in 1h 44m 07s\n- heat-functional https://zuul.opendev.org/t/openstack/build/eb361c4ee8974a04a1e61ce0e0f70641 : SUCCESS in 1h 06m 08s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/26d4b66f90f04f2ea835d37ee8a3eaca : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cdc70e35ec5bac8e46e803a148164a69b9c943a9","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-03-16 07:12:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d373459308352bd4b0e37e0c9301f23d170c112","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 07:32:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65e0fe8f0bbd423ebda60fb204cddef4 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a5cf72296494ffc95ecee594d3cb50c : SUCCESS in 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fc1202b3973f8925865b7909cbe0e718be359d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 09:07:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/832f74c40abf4242879a2548f3c7afc4 : SUCCESS in 10m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcb978c53bb34f678bd842559bb66401 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddb6115c47d04ce089242ec2c4f0bb23 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/983dfcea5aa142a98dbc8cda9506344a : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/023b16045e544c73adb6b5074a8e0135 : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bfb3a26111847ba9945410b73e79357 : SUCCESS in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3b5a18b1bc45f7bb83d601ca8fb1a3 : SUCCESS in 11m 24s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/80ed96f858154cad986778e46de8a57f : SUCCESS in 1h 52m 55s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7a7e1641a84840f89d5d2673be510fc2 : FAILURE in 57m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ce21545e62f74ed9bb7685323839e2fe : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"91517766e8f431b14061f2ef012406cec6a55c39","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-03-16 09:37:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"780fe55870f6d4d91b6899dcc90804bd45db96e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 09:57:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efe4292091424b1ba76d8d1634bf641f : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdbf9f83d2a14e4ba41d1638212810d5 : SUCCESS in 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97c465e2a5da547cd566c0f17b6bf5a9f982848a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 11:03:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06f39c435a384dcda360f7fcb9b10631 : SUCCESS in 12m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17f966869701495cb28f7ede52bc6d09 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c02f82a6d7404fba1229e4a78f8d20 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0cacd6376c74e4d9cb4f3576eac7893 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/612f4f11234c4732b9b5671ab125137b : SUCCESS in 10m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e30536aac1147478c5c4fc7eadce9ea : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05fa0db7d4344ce2bae311c47c0e864c : SUCCESS in 12m 23s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/ec97b0bda98248c59438893fde9c9666 : SUCCESS in 1h 23m 24s\n- heat-functional https://zuul.opendev.org/t/openstack/build/cfd2568b74854ae39105e6b725cd6c74 : SUCCESS in 50m 54s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/27b39a83db1d42afa9f0b2e4365b0e1a : SUCCESS in 49m 06s","accounts_in_message":[],"_revision_number":1},{"id":"14dacfc43499d8b3d4b5d390a5e7cbb2b0bfd78f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-03-17 03:52:09.000000000","message":"Topic set to 2008718","accounts_in_message":[],"_revision_number":1},{"id":"4373069e01e2c84da8715c286c36b9e5cb39df6a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-30 06:03:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f32405b53cd285b1305d46b983aae1151527823","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-08-30 11:22:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e7dfec8401beeaa3e5833b3e01c74efcd6c429ea","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-04 04:48:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1a21b0ff2e28ba5be8abe6e3f4c61987e2a6bab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 04:49:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5000f4a7d63386ca0e8f95852ad12d24f547273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 06:18:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69cc08f49844a0fbcfa3bdffb4e4c3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21b6579bc983427aa9b089c78d4b56d9 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b213bdbbae164bf2b7fb041b9238ea66 : SUCCESS in 10m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc7ea3f4a02645f0a0fd3f4956e76058 : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f966cf1b6a764ae884430dc0242f2851 : SUCCESS in 14m 18s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/cda46fb4dee44894ae7e477de2a4469d : SUCCESS in 1h 28m 00s\n- heat-functional https://zuul.opendev.org/t/openstack/build/711a886f2fac462da1f868d5f3e40c97 : SUCCESS in 1h 13m 06s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/38a668a7a0064dc8987e36c24e892800 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8f66e32a8d13cecc743464d6cfb0e6bd37ef6e3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 06:18:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b46b20e9724a9dcb033cd19cbf5b237b71778728","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 06:20:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5068f614ac37486eb9034e561fa528be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a14848c79f474c9a36fe4ae5beb368 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"520e2389d3123efc9269bbf82c6c9998b6c62564","revisions":{"520e2389d3123efc9269bbf82c6c9998b6c62564":{"kind":"REWORK","_number":1,"created":"2021-03-16 05:07:07.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/01/780701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/01/780701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/780701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/780701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/780701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/01/780701/1"}}},"commit":{"parents":[{"commit":"af0dd44dabf56fd0095714c957afe562c83f20fb","subject":"Merge \"Allow creating a Neutron port without fixed ips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/af0dd44dabf56fd0095714c957afe562c83f20fb"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-03-16 04:32:47.000000000","tz":330},"committer":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-03-16 05:06:44.000000000","tz":330},"subject":"Allow deleting user_creds when can\u0027t be decrypted","message":"Allow deleting user_creds when can\u0027t be decrypted\n\nThere are situations when the auth_encryption_key changes\nand customer wants to delete old stacks. We should allow\ndeleting those stacks.\n\nTask: #42055\nChange-Id: Ifc8c19e181902566d4f295fa979ab6869a4e0852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/520e2389d3123efc9269bbf82c6c9998b6c62564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/520e2389d3123efc9269bbf82c6c9998b6c62564"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}}]}],"submit_requirements":[]}
