)]}'
{"id":"openstack%2Fheat~784409","triplet_id":"openstack%2Fheat~stable%2Fwallaby~Icc90a921ec67e49aec2c4acfad72235c57c78421","project":"openstack/heat","branch":"stable/wallaby","hashtags":[],"change_id":"Icc90a921ec67e49aec2c4acfad72235c57c78421","subject":"Don\u0027t update status for replaced resource","status":"MERGED","created":"2021-04-02 03:46:41.000000000","updated":"2021-04-07 07:19:16.000000000","submitted":"2021-04-07 07:17:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784409","cherry_pick_of_change":784199,"cherry_pick_of_patch_set":2,"meta_rev_id":"3a900f1f8b482f6feda7f71831192bbdf8b7e9b2","_number":784409,"virtual_id_number":784409,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 07:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-04-07 05:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-04-07 05:31:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-04-07 05:32:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 04:19:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-02 05:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 05:31:45.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"}],"messages":[{"id":"2ac9929bc81648460f4f3cd54f87f31d87deb263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-04-02 03:46:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bb43b296e335d675b1832700f7d06cb4b5b175b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 04:19:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82a43e35945543a097cbc26ba2a2a1b0 : SUCCESS in 14m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf0c5d61509145babd17591be44a0f52 : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bb171130567e2c0aba7fc86e0085438dae69142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 05:12:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/870f650f49d94c3f9faace35d3ff86ba : SUCCESS in 8m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6247bbc8b71c43a3983ff46142436bbc : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffed26e7ae4844569786c21d03503a9b : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6789cad633914cc081b63e1773bfd9f5 : SUCCESS in 9m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f85ab060e0d24050a39a73a49984a511 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78392f4c29544a70af28f0482e956685 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60785727686496b9762aea389414d9e : SUCCESS in 15m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/011302def1074343ae9d4fb4c8d4696e : SUCCESS in 1h 21m 53s\n- heat-functional https://zuul.opendev.org/t/openstack/build/284653905e8849099e1180bf122eff77 : FAILURE in 1h 11m 38s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/81b2b5c07fd243b9a89fbaddee4b1f21 : FAILURE in 1h 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"88088d1b4d1d63417237f1c62d6f1de3ee122005","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-04-02 05:36:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9e273bb6baf20a3a7a53179da85746b9ba2ada5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 05:57:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bcd72e01e474694b9564fa8e7743f87 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5085e2d849c24c18930211db2363cebc : SUCCESS in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d58d123c304897f241d55ba5d796a2bee788be51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 07:15:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e828b43c00cd4d4c993e9abdcd2e4d6a : SUCCESS in 10m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8adc5065121e47e6b203fbb0a04b812c : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a060c2c78b445fa956bab2415e5ece1 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cee8906fb1a4a94bf8551e071f5130e : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27170415cbf04ec48bda266ab8cf0249 : SUCCESS in 8m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9180236060b4cbc8bb89a7c81a32ebf : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc7e7c78a3640b1a43928035d2e5f48 : SUCCESS in 12m 27s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6832559fc5c6484fa96efa41cfd55e37 : SUCCESS in 1h 36m 19s\n- heat-functional https://zuul.opendev.org/t/openstack/build/469afe4f7044401f8a5fb621c0c9b962 : SUCCESS in 1h 00m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a0c2aa4bd98548a2926139e6a6772892 : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":2},{"id":"36da288a378599f852cd660c3b036a165d06cd91","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-04-07 05:31:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f2e1c04518afa1e3b6acb913cbb4d61676f4fde6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-04-07 05:32:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3fe029cb191261f42780c34960d82fb4c37bbf47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 05:32:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd3cf1ad42e439c80ee60f3346c811a859ef6f30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 07:17:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b65ffe011d994da4b84b0045cf1cf83a : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e857f28d0b4338b144aedd79fc2c5c : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d85f7f932b4f5189a615d244a59366 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9db4d1de9d4b46b396e14d2e493f2eda : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac27f6eabf2478aaed42e525a3c06ae : SUCCESS in 10m 09s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/367a3b8f4e054a44bec56ff4b078e70a : SUCCESS in 1h 29m 18s\n- heat-functional https://zuul.opendev.org/t/openstack/build/72c483a968284c179b4f8beb26c64893 : SUCCESS in 58m 24s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/e9ce23979bbf4fe3b18d0f009911301f : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":2},{"id":"aab31545dffa3ba9151bb5478f06cb671f6da788","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 07:17:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3a900f1f8b482f6feda7f71831192bbdf8b7e9b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 07:19:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868276034c9c43bfbe259f87f20de229 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7af71497ed15882a0f2a543cd8dd83c60b781f95","revisions":{"d7dc3a391afbb6a5c0cc78bda266824ca1d9247d":{"kind":"REWORK","_number":1,"created":"2021-04-02 03:46:41.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/09/784409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/09/784409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/784409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/784409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/784409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/09/784409/1"}}},"commit":{"parents":[{"commit":"af0dd44dabf56fd0095714c957afe562c83f20fb","subject":"Merge \"Allow creating a Neutron port without fixed ips\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/af0dd44dabf56fd0095714c957afe562c83f20fb"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-04-01 04:49:31.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2021-04-02 03:46:41.000000000","tz":0},"subject":"Don\u0027t update status for replaced resource","message":"Don\u0027t update status for replaced resource\n\nWith I04e7ad90944c2d03ce0e59ba16af9d60d6e01222 we allowed\nupdate of previously-existing resource if replacement\ncreation has failed during last update. However, we force\nupdate the status of the existing resource to \u0027COMPLETE\u0027.\nTherefore, in the next update if properties/type has not\nchanged for the resource it won\u0027t try to update/replace\nthe previously existing resource resulting in false positive\nof stack updated successfully.\n\nTask: 42194\nChange-Id: Icc90a921ec67e49aec2c4acfad72235c57c78421\n(cherry picked from commit bfbaab107d709c37548a22239b952c6221b9ac85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d7dc3a391afbb6a5c0cc78bda266824ca1d9247d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d7dc3a391afbb6a5c0cc78bda266824ca1d9247d"}]},"branch":"refs/heads/stable/wallaby"},"7af71497ed15882a0f2a543cd8dd83c60b781f95":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-02 05:36:58.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/09/784409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/09/784409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/784409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/784409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/784409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/09/784409/2"}}},"commit":{"parents":[{"commit":"0bffa5c2e0e2f12264afebcadda0ec9617e9ed69","subject":"Assume interface detach called for port not found","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0bffa5c2e0e2f12264afebcadda0ec9617e9ed69"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-04-01 04:49:31.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2021-04-02 05:36:58.000000000","tz":0},"subject":"Don\u0027t update status for replaced resource","message":"Don\u0027t update status for replaced resource\n\nWith I04e7ad90944c2d03ce0e59ba16af9d60d6e01222 we allowed\nupdate of previously-existing resource if replacement\ncreation has failed during last update. However, we force\nupdate the status of the existing resource to \u0027COMPLETE\u0027.\nTherefore, in the next update if properties/type has not\nchanged for the resource it won\u0027t try to update/replace\nthe previously existing resource resulting in false positive\nof stack updated successfully.\n\nTask: 42194\nChange-Id: Icc90a921ec67e49aec2c4acfad72235c57c78421\n(cherry picked from commit bfbaab107d709c37548a22239b952c6221b9ac85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/7af71497ed15882a0f2a543cd8dd83c60b781f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/7af71497ed15882a0f2a543cd8dd83c60b781f95"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
