)]}'
{"id":"openstack%2Fheat~784582","triplet_id":"openstack%2Fheat~stable%2Fussuri~Icc90a921ec67e49aec2c4acfad72235c57c78421","project":"openstack/heat","branch":"stable/ussuri","hashtags":[],"change_id":"Icc90a921ec67e49aec2c4acfad72235c57c78421","subject":"Don\u0027t update status for replaced resource","status":"MERGED","created":"2021-04-04 08:10:20.000000000","updated":"2021-04-07 14:53:30.000000000","submitted":"2021-04-07 14:50:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784582","cherry_pick_of_change":784199,"cherry_pick_of_patch_set":2,"meta_rev_id":"d6075c4ff6eb8b15406bb1bb8bef1f877a7e47a3","_number":784582,"virtual_id_number":784582,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 14:50:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-04-07 13:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-04-07 05:31:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-04-07 13:07:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-04 09:26:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 05:31:51.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"}],"messages":[{"id":"6d3cf6dd9b25029a80253e3732065f028e134e66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-04-04 08:10:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3e2bab720c5a5190cad80e34ccb9cc5d81297f68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 09:26:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b9b72b1c73b4b80b5447634430eb1f2 : SUCCESS in 12m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3bdf68a2b5456d8dca5b7ff3ef2e76 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d683cf7047794d9794f7334f3aa2fce2 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3989f7724b4c4fae7abd14c0e372a7 : SUCCESS in 9m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fe57611125b4181b0ee03c6674a7d2c : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016e947df42a42379176d7fdea960a89 : SUCCESS in 9m 50s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d16d6327cbdd4a3bbf9087c5706f7281 : SUCCESS in 1h 12m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/845739b63b8a4958ac1186c3f74884ad : SUCCESS in 1h 14m 31s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7633d7e56aba47ffa24fc65859d85a6b : SUCCESS in 48m 22s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/e3e28460657c461bb323c78a87c887a1 : SUCCESS in 46m 44s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/de06a48214214e969181cf4411eb1ecf : SUCCESS in 56m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50f03fbde7fe9b32420e51bb29d89de3d66f54f9","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2021-04-07 05:31:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b27b0ca4a0ee152e84c3ebb670093f8c150d9a66","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-04-07 13:07:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c043f4d274cfafd9c1f4d269ae448214d25d359a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 13:07:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3542343cda49a21ae3dd61084817a567218a6456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 14:50:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3036ec284eff4c979b0f1d3cc8f4fd0f : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de5a12a2bd64746bfe40bf2c7555015 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac7fa489631b4cca8c7402c2d58ee1c4 : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7179fcead53b41e0a46da260ba9dc725 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e015707dc3c4483b82b5efcc7ad83ee : SUCCESS in 12m 52s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d9af912d9aea409cb65285b41dc1166f : SUCCESS in 1h 24m 35s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9900e4b971b34216851ca6a286ab7639 : SUCCESS in 1h 40m 24s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7de0057deb1d4164adfd81412e53160c : SUCCESS in 1h 03m 01s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/196e13167faf40f4b2d62af48fc99bc4 : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fd50826ab47aeada75722d8c19e7ff93af6befa6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 14:50:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6075c4ff6eb8b15406bb1bb8bef1f877a7e47a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 14:53:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53314a2cca8f47f2b7f81612c31e6c10 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e95a7fd21deea111d97b9e8e8ecff431091100d9","revisions":{"e95a7fd21deea111d97b9e8e8ecff431091100d9":{"kind":"REWORK","_number":1,"created":"2021-04-04 08:10:20.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/82/784582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/82/784582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/784582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/784582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/784582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/82/784582/1"}}},"commit":{"parents":[{"commit":"49ef181bb229ded1131582a64fd4cba9be6fc953","subject":"Add SOURCE_IP_PORT to LB_ALGORITHM allowed values","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/49ef181bb229ded1131582a64fd4cba9be6fc953"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-04-01 04:49:31.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2021-04-04 08:10:20.000000000","tz":0},"subject":"Don\u0027t update status for replaced resource","message":"Don\u0027t update status for replaced resource\n\nWith I04e7ad90944c2d03ce0e59ba16af9d60d6e01222 we allowed\nupdate of previously-existing resource if replacement\ncreation has failed during last update. However, we force\nupdate the status of the existing resource to \u0027COMPLETE\u0027.\nTherefore, in the next update if properties/type has not\nchanged for the resource it won\u0027t try to update/replace\nthe previously existing resource resulting in false positive\nof stack updated successfully.\n\nTask: 42194\nChange-Id: Icc90a921ec67e49aec2c4acfad72235c57c78421\n(cherry picked from commit bfbaab107d709c37548a22239b952c6221b9ac85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e95a7fd21deea111d97b9e8e8ecff431091100d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e95a7fd21deea111d97b9e8e8ecff431091100d9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
