)]}'
{"id":"openstack%2Fheat~787983","triplet_id":"openstack%2Fheat~stable%2Fvictoria~Iaa2c05b4155d93f44de60b6f98a69450c1512817","project":"openstack/heat","branch":"stable/victoria","hashtags":[],"change_id":"Iaa2c05b4155d93f44de60b6f98a69450c1512817","subject":"Preserve order in list_concat_unique","status":"MERGED","created":"2021-04-26 14:50:47.000000000","updated":"2021-04-27 06:58:55.000000000","submitted":"2021-04-27 06:56:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787983","cherry_pick_of_change":787837,"cherry_pick_of_patch_set":1,"meta_rev_id":"a7cbfe68a25480863a54f95999f2c474a5fcaf50","_number":787983,"virtual_id_number":787983,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 06:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":2,"date":"2021-04-26 15:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":1,"date":"2021-04-26 15:19:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 15:19:05.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2021-04-26 16:27:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39c0759a8a1349740b9a9c0394896b53e3a7cf1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-04-26 14:50:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"1b12f6371ef17181d047c36dda32467c3cfaa7ec","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-04-26 15:19:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThe old version was sufficiently non-functional that I think this backport is justified.\nMerged in Wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"8854aedf9baf749f7f02774ee29cd4797f6a1973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:27:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/856afd5e73ed4f45853be395468447c6 : SUCCESS in 9m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a45f396f0207424db27e75cc7bc19232 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/976c6354787341daa187ea103a028ce8 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a6bcabcaf26489aa96736ad7fad1f8e : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f579ea7f18d44e0bbbf0b3a7b62aec59 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0212dc177e494844b7c4829585d8105f : SUCCESS in 12m 15s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ebd459b4c1e048e1bb1e1c53ec8736e6 : SUCCESS in 1h 20m 30s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/363f672bd8934cf2852a942488aad56a : SUCCESS in 1h 35m 02s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7839a65e112242c99eacf7dd93277062 : SUCCESS in 1h 04m 13s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/b8279f41eb264a7597f363246624b1b3 : SUCCESS in 1h 10m 01s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/296e0b0f046c44eab6d38842689c4f88 : SUCCESS in 58m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61e7898e7d91f4467f2322afeaee6d603199d18e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:27:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d81a1934276e7691f57ff1e310b71f3ad833794","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:56:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3f9c7b17cf341d293f138e999eaf02c : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/510c17be23c9436b8f8e170b090ddcc3 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/908d421807364cd7a84a6fb15b22b8a4 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a951acd9a596413384be2a6906f58f1c : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a65f264aa3471ca29705ab134316f7 : SUCCESS in 13m 20s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/6421d842b9894f74bdd29c3a75b55992 : SUCCESS in 1h 19m 09s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/2f5ef0384a714eb7b80c9039d6ba4187 : SUCCESS in 1h 27m 24s\n- heat-functional https://zuul.opendev.org/t/openstack/build/a4c285c1f30e437cbb34e890bfb241dd : SUCCESS in 56m 59s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7c2f6898586f45c2874ffb062f3b6d3f : FAILURE in 55m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2e1954b17301c9a0e61ef8ecf037a0393f556af6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-04-27 03:56:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c9f2b169d9503c571da82b5c72a6da72fdbd033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 05:24:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/994f595840d14b5689b5be0638f65ffb : SUCCESS in 7m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19568d7336034ade9650523ae0cab228 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79129a9f0984ba498d7b49b62eea1f8 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1236eb3a8707458ca212ff563265a60d : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72a892b1cef54be3976fad00d04f0c10 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0a106674254b238b2e1b5386b3cf73 : SUCCESS in 12m 34s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d45ab941bd25405bad5efd47e999be99 : SUCCESS in 1h 21m 41s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/120d295cf0204528abf7125c4d306388 : SUCCESS in 1h 26m 47s\n- heat-functional https://zuul.opendev.org/t/openstack/build/786d8d7d389a4ddfa5bb56470e17ee15 : SUCCESS in 58m 31s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/bbbec1ef94a84ce5a0adbc5fae87104f : SUCCESS in 57m 20s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/2818a9ee65264bd6affe7b27ccae782a : SUCCESS in 59m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e1d953c6bb7b5882ad8695863e8269e416581c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 05:25:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"529bbc1210891d7b0e8c54306bc38e2a10bdcbed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 06:56:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc236f728c7f4dc183410c6181109195 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f08dd350ee2410b8b232ef12756d13d : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fab87667c3a14bdcb9839bdd5a4fff57 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d657945a19814a37bc6866a7dc11bf3d : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98cfcdb5f99469c981608d0d1f30885 : SUCCESS in 11m 18s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/add7bc12068c4e199fa2191f5043b720 : SUCCESS in 1h 16m 51s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/fe186a869f914704b4af7e8af731764d : SUCCESS in 1h 30m 21s\n- heat-functional https://zuul.opendev.org/t/openstack/build/8210707ac5d347f89e04654608bc3885 : SUCCESS in 1h 05m 58s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9374c8b73a6141d8b9f6c9e429ed219c : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0cb0ed8dc7103053015099e9eb6b727d775eb391","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 06:56:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7cbfe68a25480863a54f95999f2c474a5fcaf50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 06:58:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d634e5d7294817b9de9f034351cfaf : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0833daa3f77dd2267a6068eeafcf9d9af8da1a17","revisions":{"0833daa3f77dd2267a6068eeafcf9d9af8da1a17":{"kind":"REWORK","_number":1,"created":"2021-04-26 14:50:47.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/83/787983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/83/787983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/83/787983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/83/787983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/83/787983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/83/787983/1"}}},"commit":{"parents":[{"commit":"9181439680347c4f0bddd9a700d45d7dcd6f7969","subject":"Merge \"Assume interface detach called for port not found\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9181439680347c4f0bddd9a700d45d7dcd6f7969"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-04-23 03:06:37.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2021-04-26 14:50:47.000000000","tz":0},"subject":"Preserve order in list_concat_unique","message":"Preserve order in list_concat_unique\n\nAlso, we should not modify a list when iterating\nover it.\n\nTask: 42359\nCloses-Bug: #1925373\nChange-Id: Iaa2c05b4155d93f44de60b6f98a69450c1512817\n(cherry picked from commit a8f8528d1ea1952cf76bd09fd27cbd65a978af65)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0833daa3f77dd2267a6068eeafcf9d9af8da1a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0833daa3f77dd2267a6068eeafcf9d9af8da1a17"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
