)]}'
{"id":"openstack%2Fheat~794800","triplet_id":"openstack%2Fheat~stable%2Fussuri~I48ea3cb107598b7406355d8316d9104c39633e58","project":"openstack/heat","branch":"stable/ussuri","topic":"missing-tgtermes.sty","hashtags":[],"change_id":"I48ea3cb107598b7406355d8316d9104c39633e58","subject":"[stable-only] Add missing font for PDF generation","status":"ABANDONED","created":"2021-06-04 12:47:50.000000000","updated":"2023-03-16 03:33:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fcb556deff2bc7ce93a4ad3616b6a6dde740bf4e","_number":794800,"virtual_id_number":794800,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-04 15:24:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 15:24:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 09:13:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2021-06-07 09:17:32.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"ee874d4bf62189464f55763ce7d684358a68fc21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-04 12:47:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cb6e8f631e5b69d6aa0780668e1195e99d3c820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 15:24:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d09927e95d7a48d3853da15317622e82 : SUCCESS in 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/524c342adaa148c1ba4f7feda51dcaae : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6ccc6e0211491aae63fe3174ebba0e : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efa1239a0887432bbce146b98325bb41 : SUCCESS in 9m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6fd0de5d3bb844ac980e1a35e7e1af6f : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338e4af74d0642ef8409c9d4ba2b9836 : SUCCESS in 13m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c6f1a80ddf544459342534088e1badf : SUCCESS in 10m 06s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/dabe6cefa8c843dea115c7f4c6e1c2f0 : SUCCESS in 1h 26m 13s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4a019bf1b7244f54aea9f5c874cb1813 : SUCCESS in 1h 36m 44s\n- heat-functional https://zuul.opendev.org/t/openstack/build/88ef64df976d440d8ca58218cf5fd876 : SUCCESS in 1h 08m 03s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ea05b5fcf6df4d31ae2ee764c3cc3aed : SUCCESS in 1h 00m 54s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/c2a45f46df0d4b73a09228fb23739cf8 : SUCCESS in 1h 04m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ad75aca2415fecbade193fb88c6464b28d28f81","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-07 09:16:27.000000000","message":"Patch Set 1:\n\nThis patch fixes stable/ussuri gate (and thus periodic-stable job [1]).\n\n[1] http://lists.openstack.org/pipermail/openstack-stable-maint/2021-June/thread.html#43631","accounts_in_message":[],"_revision_number":1},{"id":"fcb556deff2bc7ce93a4ad3616b6a6dde740bf4e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-16 03:33:46.000000000","message":"Abandoned\n\nWe agreed with EOLing train and ussuri, so I\u0027ll abandon this change.\n\nIn case you have any concern please let us know in https://review.opendev.org/c/openstack/releases/+/871857 .\n\nNote that we require some volunteers to fix CI jobs in these old branches in case anyone wants to keep these open.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84b73ea2fb41894519db8714e0ae175a837ffd91","revisions":{"84b73ea2fb41894519db8714e0ae175a837ffd91":{"kind":"REWORK","_number":1,"created":"2021-06-04 12:47:50.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/00/794800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/00/794800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/00/794800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/00/794800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/00/794800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/00/794800/1"}}},"commit":{"parents":[{"commit":"b9a3395978bf60d3ce769b8fcedc60b7127f404f","subject":"Preserve order in list_concat_unique","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b9a3395978bf60d3ce769b8fcedc60b7127f404f"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 12:29:07.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-06-04 12:47:03.000000000","tz":120},"subject":"[stable-only] Add missing font for PDF generation","message":"[stable-only] Add missing font for PDF generation\n\nopenstack-tox-docs job started to fail with the following error:\n\n! LaTeX Error: File `tgtermes.sty\u0027 not found.\n\nThis patch adds the required font package to bindep doc profile.\n\nChange-Id: I48ea3cb107598b7406355d8316d9104c39633e58\n(cherry picked from commit c9d824bac0b892090f56acb6e970c9e93bcdd1e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/84b73ea2fb41894519db8714e0ae175a837ffd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/84b73ea2fb41894519db8714e0ae175a837ffd91"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
