)]}'
{"id":"openstack%2Fheat~803682","triplet_id":"openstack%2Fheat~stable%2Fvictoria~I36421e9859a266f0278c1b5f2acb4ebbacbfca82","project":"openstack/heat","branch":"stable/victoria","topic":"bug/1740567","hashtags":[],"change_id":"I36421e9859a266f0278c1b5f2acb4ebbacbfca82","subject":"Update get_hosts to use available API","status":"MERGED","created":"2021-08-06 13:11:19.000000000","updated":"2021-09-13 09:53:38.000000000","submitted":"2021-09-13 09:51:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803682-bug/1740567","cherry_pick_of_change":682061,"cherry_pick_of_patch_set":27,"meta_rev_id":"a5c12033451f1b5505beb6953f7102692e4e60e5","_number":803682,"virtual_id_number":803682,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-13 09:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-09-13 08:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2021-09-13 09:53:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-09-13 08:23:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-06 13:11:19.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-08-06 13:12:23.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-08-06 13:12:23.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2021-08-06 13:12:23.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-08-06 14:01:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 09:51:00.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"}],"messages":[{"id":"c880611cc56a1d07e1662232048a490f0a5e6c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-06 13:11:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bc1beb10a2b61424e53591bee39c768e44b4ac80","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-06 13:11:50.000000000","message":"Topic bug/1740567-stable/victoria removed","accounts_in_message":[],"_revision_number":1},{"id":"bc19e92783d535719d7db6e84eea1c4be5ae6c63","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-06 13:11:52.000000000","message":"Topic set to bug/1740567","accounts_in_message":[],"_revision_number":1},{"id":"c8d4f186ac1c0e58828f1e8f4ede33f507c0907a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-06 13:30:09.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks valid to backport, clean cherry pick from master (Wallaby), where the patch has merged. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"3fdbd7a23861133fd00209c8283f5f5a15994179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 14:01:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6043ebf190954d218e02eff5ff149327 : SUCCESS in 10m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b8cb6453f604fb5b8ed2fb19a2aa454 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba3bbf1da5d456d99ef008033908338 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4eb1f09f91642b2ab772fa7b24198d4 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/342ecfeadc2047389e6d12128f3ec56a : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bcfb49b19c446c285b981a8b6f5da72 : SUCCESS in 11m 14s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/81590c8bcc05477490a3212c85aa2da4 : FAILURE in 36m 55s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4f1722ed0a9046fa8c007534de2f7c11 : FAILURE in 45m 58s\n- heat-functional https://zuul.opendev.org/t/openstack/build/4475c4ed1a67462abbaef611b2bfe86e : FAILURE in 39m 54s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/999c8866b8314d9eb3e7b55bf9fdd269 : FAILURE in 39m 56s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/4f50774e98564e02b2cc0ee27d56ed34 : FAILURE in 39m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"455234da2322a78799ea574354a77f0ab587f594","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-06 16:23:58.000000000","message":"Patch Set 1:\n\nThe jobs are failing because the Fedora 32 image is not reachable on mirror site [1]. I don\u0027t know whether this is some temporary problem or the Fedora 32 image was removed from the opendev mirror permanently.\n\n[1] logs:\n++ functions:upload_image:99                :   wget --progress\u003ddot:giga -c http://mirror-int.iad.rax.opendev.org/fedora/releases/32/Cloud/x86_64/images/Fedora-Cloud-Base-32-1.6.x86_64.qcow2 -O /opt/stack/old/devstack/files/Fedora-Cloud-Base-32-1.6.x86_64.qcow2\n--2021-08-06 13:50:15--  http://mirror-int.iad.rax.opendev.org/fedora/releases/32/Cloud/x86_64/images/Fedora-Cloud-Base-32-1.6.x86_64.qcow2\nResolving mirror-int.iad.rax.opendev.org (mirror-int.iad.rax.opendev.org)... 10.208.224.52\nConnecting to mirror-int.iad.rax.opendev.org (mirror-int.iad.rax.opendev.org)|10.208.224.52|:80... connected.\nHTTP request sent, awaiting response... 404 Not Found\n2021-08-06 13:50:15 ERROR 404: Not Found.","accounts_in_message":[],"_revision_number":1},{"id":"6bf7a6da685fd0954dcb1fbc82c834b59b136768","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-18 09:41:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5812717108cf429b0714308bb705bfeaadf4b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:30:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8643047ab444411d82fd829d4fcf52c1 : SUCCESS in 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8612f87cc2e04cf598a8573ab367bf29 : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c314302664b343538dcd4346f46d4748 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6dfe66f87804be2adeea397f0ce69c5 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70cef9ed060a45728e4bcee052be72d0 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb787b329624164b7058bc4e5b09709 : SUCCESS in 11m 34s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/8b8c4139473144c5be195b55cfbe0567 : FAILURE in 33m 15s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6d2efec8374d43dfa1ad764f55f2dac5 : FAILURE in 45m 11s\n- heat-functional https://zuul.opendev.org/t/openstack/build/5c83631b00574f8484ace0d0211a1b0e : FAILURE in 37m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/3b22a63e7c2646a7a29b0b77ebbbe48f : FAILURE in 37m 28s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/36f11ff96b854a9da552ee8fe97c810a : FAILURE in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c76f3b3816156cccee5ee8323e7409f17efd042","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-01 12:47:59.000000000","message":"Patch Set 1:\n\nrecheck based on the positive test result in https://review.opendev.org/c/openstack/heat/+/805005/ I assume that the missing fedora image case is resolved now","accounts_in_message":[],"_revision_number":1},{"id":"21ae463891dceeacd586215701e35ac82a036748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 14:21:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/668132dcdc6444508b6c920a662774a0 : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33e0ff0697044c84a17e1ce47633fb79 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515812469706463fbe609d7c98797530 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c2ae22ecd6e485aaf4b4ef739727320 : SUCCESS in 11m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e744e45f2884ea4a08bf3beba431369 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15f90b2be37f4bf783ad20f113c0252d : SUCCESS in 14m 41s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/98f98e53b424445b9bbcf0f58d74cd6e : SUCCESS in 1h 20m 00s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/7e9b79fbdab740f986b7ee182547c341 : SUCCESS in 1h 27m 00s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/25370047ceac4f9580425f77f2a36b7f : SUCCESS in 59m 09s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7f97238adfed490688b1b9adedce06df : SUCCESS in 1h 07m 55s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/98965e73fba946bfb871dbe9a0df7000 : SUCCESS in 1h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"640e2800bfea7f58b270ddbcee6ea67b58b9de40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-09-10 09:49:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"82cb35a3a9770f15accd47d8601b57802c692371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 11:21:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16de9838a9e14373957c216c4c79819e : SUCCESS in 11m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e837acaf5b9b493ea38cb68601da7469 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84efaed10dcf4991804a4232945609eb : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47a3e6d5b6b94c00975f9b1d022f9719 : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ca47d349d034788b214af7f236b5aac : SUCCESS in 10m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b298c31246844d9ebb5aad5a0cfed2b9 : SUCCESS in 11m 50s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/939a7012853748d2b03aec45fc268851 : SUCCESS in 1h 26m 03s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9b5a87da3a7f4624b865f7b01c816b9e : SUCCESS in 1h 27m 38s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/86887c51d4aa422cbe0663bf875d7ad9 : SUCCESS in 1h 06m 32s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d79961808d5c406e93652f71cbb12402 : SUCCESS in 1h 04m 53s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/2be2fa7e6f124e5c83b38345eb28dd27 : SUCCESS in 1h 01m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"673ec99f86b016aee3faa5f9a11f6e7ebb34c2b7","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-09-13 08:23:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f8ceb5449fb41a5b7c6a11529cef4565d91eebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:24:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4cda2bf5a9f0eb5f1ddc9f258e2ec791c0f0d5d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 09:51:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d288d2a0c24e468181540a2e94f3eb87 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349a9aeb6228419f99a06f6552750c11 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e32034d1e7549aaae0117e28bba2f6e : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f9d4a3601ab457ab32e1d6531a04029 : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b33d0f2ce88e41ae80c593fdd33597b9 : SUCCESS in 13m 23s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d352925bf97c429b823423186d7c3526 : FAILURE in 1h 08m 40s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0704727468db404abcbf9e36b461e315 : SUCCESS in 1h 25m 58s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/0603f9bd4bc84fb6bdd17aa723f5f2d3 : SUCCESS in 59m 07s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/bdec15d400bb43fba4ab4d90feea9b1b : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7231316a7a81d60bd143694c23704ce768b63218","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 09:51:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5c3609d86d31071624c7c7b5d08b0c3ff322499b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 09:53:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/594b633f47444b4a82d9d00263dc9089 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40a3494c86c36fb4987f8b9a7c296f9411d1263b","revisions":{"a435b315a3f10d1bc84ed15efbe44be0b9bd11be":{"kind":"REWORK","_number":1,"created":"2021-08-06 13:11:19.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/82/803682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/82/803682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/803682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/803682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/803682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/82/803682/1"}}},"commit":{"parents":[{"commit":"0833daa3f77dd2267a6068eeafcf9d9af8da1a17","subject":"Preserve order in list_concat_unique","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0833daa3f77dd2267a6068eeafcf9d9af8da1a17"}]}],"author":{"name":"David J Peacock","email":"david.j.peacock@gmail.com","date":"2019-09-13 14:08:02.000000000","tz":-240},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-06 13:11:19.000000000","tz":0},"subject":"Update get_hosts to use available API","message":"Update get_hosts to use available API\n\nos-hosts was deprecated in nova quite some time ago.  This patch updates\nget_hosts to use the available os-hypervisors functions.\n\nChange-Id: I36421e9859a266f0278c1b5f2acb4ebbacbfca82\nStory: 2006584\nTask: 36708\nbz: 1740567\n(cherry picked from commit aab68ced6a84c9fa39adb2cb9bb19165e699ff3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a435b315a3f10d1bc84ed15efbe44be0b9bd11be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a435b315a3f10d1bc84ed15efbe44be0b9bd11be"}]},"branch":"refs/heads/stable/victoria"},"40a3494c86c36fb4987f8b9a7c296f9411d1263b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-10 09:49:42.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/82/803682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/82/803682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/803682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/803682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/82/803682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/82/803682/2"}}},"commit":{"parents":[{"commit":"d5d8f74e6d9642c1df6a48dadb6ab828cf2ffe05","subject":"Merge \"Detect EL8 platform-python\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d5d8f74e6d9642c1df6a48dadb6ab828cf2ffe05"}]}],"author":{"name":"David J Peacock","email":"david.j.peacock@gmail.com","date":"2019-09-13 14:08:02.000000000","tz":-240},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-10 09:49:42.000000000","tz":0},"subject":"Update get_hosts to use available API","message":"Update get_hosts to use available API\n\nos-hosts was deprecated in nova quite some time ago.  This patch updates\nget_hosts to use the available os-hypervisors functions.\n\nChange-Id: I36421e9859a266f0278c1b5f2acb4ebbacbfca82\nStory: 2006584\nTask: 36708\nbz: 1740567\n(cherry picked from commit aab68ced6a84c9fa39adb2cb9bb19165e699ff3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/40a3494c86c36fb4987f8b9a7c296f9411d1263b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/40a3494c86c36fb4987f8b9a7c296f9411d1263b"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
