)]}'
{"id":"openstack%2Fheat~809229","triplet_id":"openstack%2Fheat~stable%2Frocky~Ifd1b570a7288c8c3ef322b6ad931d3ac9dcc354f","project":"openstack/heat","branch":"stable/rocky","topic":"cleanup-featuresets-from-old-branches-stable/rocky","hashtags":[],"change_id":"Ifd1b570a7288c8c3ef322b6ad931d3ac9dcc354f","subject":"Remove jobs corresponds to obselete featuresets","status":"ABANDONED","created":"2021-09-16 09:11:25.000000000","updated":"2023-01-23 16:54:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":809357,"cherry_pick_of_patch_set":1,"meta_rev_id":"929a95871e026179ea178466e4d2005e7484cc08","_number":809229,"virtual_id_number":809229,"owner":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-16 10:48:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 09:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa7e25761c11d9ed52564c3904f86c7ce3ecf7b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 09:11:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.\n\nThe following files contain Git conflicts:\n* .zuul.yaml","accounts_in_message":[],"_revision_number":1},{"id":"3ac34b75364b82454c35a7a27525468c40c06d14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 09:11:55.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/heat on branch stable/rocky.  The error was:\n\n  while scanning a simple key\n  in \"openstack/heat/.zuul.yaml@stable/rocky\", line 130, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/heat/.zuul.yaml@stable/rocky\", line 132, column 9","accounts_in_message":[],"_revision_number":1},{"id":"422ca0e4581f74cfca0d9754dbdaf46eaba236ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 09:12:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9d425b2bf4dd4d1dd187631e2ff662704f369e6d","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 09:12:49.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"00c88295217b884b2263065f5445523226df8e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:48:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd0b2fe4559b43079376470b9c39a2f8 : SUCCESS in 8m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c385696daca44b7b4c8d7d5a1fdf6b1 : SUCCESS in 10m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7567628f615a460585d0386615dad7b3 : SUCCESS in 10m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5f6422b1c6ef4e6eb1820f0b00dd2156 : SUCCESS in 10m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e646beb38a264eee8f2f549371971804 : FAILURE in 4m 34s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d3dbbc415527433186930eb4ac68ec57 : SUCCESS in 1h 15m 28s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/75762c01b85a44f696edfe3dcb88d08a : SUCCESS in 1h 26m 16s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/25b3772c66c840f8bfe3639a90fcc4be : FAILURE in 31m 38s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/ce4dd8b377be4cf5a734c01bc382b9a7 : FAILURE in 32m 38s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/dc0e2645db424912bf02b223a22808f8 : FAILURE in 32m 21s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py35 https://zuul.opendev.org/t/openstack/build/9df0206d143b42f791b83686855b29ab : FAILURE in 34m 02s","accounts_in_message":[],"_revision_number":2},{"id":"849ea615a3b416036f95bc0545827d5281329662","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2021-09-16 11:17:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"85567b9980defe00ecefe643297debd26e958128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 12:36:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0cc1078c13040e9b30edd210d734345 : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e07d0996fea34701870ca5a8b98509b9 : SUCCESS in 11m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1421658d455f43d7a8cf8cb45bfd72e6 : SUCCESS in 8m 30s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19177cf20ba049d6a10f521c8f9a1f3d : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ca58cc120a14d7d89197ef34b6419cf : FAILURE in 4m 53s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/a2e1312c9301425783e5726ce88e2f33 : SUCCESS in 1h 12m 21s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/b460367d25f449e79d66fe92f11b2238 : SUCCESS in 1h 13m 01s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/ff1a113860984d609aae799e3e81d66f : FAILURE in 31m 48s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/5d44b3e1d8c2491abf17c0bfd8edfbc3 : FAILURE in 27m 46s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/7a944c95ed7a43deafb51d572e7a628e : FAILURE in 31m 27s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py35 https://zuul.opendev.org/t/openstack/build/b41201ac42fa4cc2bd4b645360654f0f : FAILURE in 36m 49s","accounts_in_message":[],"_revision_number":2},{"id":"929a95871e026179ea178466e4d2005e7484cc08","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-01-23 16:54:14.000000000","message":"Abandoned\n\nstable/rocky of heat has transitioned to End of Life. To be able to delete the branch all patches need to be abandoned.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0449fb882a133c3b05264da4961c5dce26dcefcc","revisions":{"b43aea1d260054dbac7785493751845db8bb8ecf":{"kind":"REWORK","_number":1,"created":"2021-09-16 09:11:25.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/29/809229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/29/809229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/29/809229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/29/809229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/29/809229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/29/809229/1"}}},"commit":{"parents":[{"commit":"b122028baea73a11f7459f2d378485e7deabcffd","subject":"Merge \"Check external resources after creation\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b122028baea73a11f7459f2d378485e7deabcffd"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 09:09:08.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 09:11:25.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Ifd1b570a7288c8c3ef322b6ad931d3ac9dcc354f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b43aea1d260054dbac7785493751845db8bb8ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b43aea1d260054dbac7785493751845db8bb8ecf"}]},"branch":"refs/heads/stable/rocky"},"0449fb882a133c3b05264da4961c5dce26dcefcc":{"kind":"REWORK","_number":2,"created":"2021-09-16 09:12:44.000000000","uploader":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"ref":"refs/changes/29/809229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/29/809229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/29/809229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/29/809229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/29/809229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/29/809229/2"}}},"commit":{"parents":[{"commit":"b122028baea73a11f7459f2d378485e7deabcffd","subject":"Merge \"Check external resources after creation\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b122028baea73a11f7459f2d378485e7deabcffd"}]}],"author":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 09:09:08.000000000","tz":330},"committer":{"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","date":"2021-09-16 09:12:37.000000000","tz":0},"subject":"Remove jobs corresponds to obselete featuresets","message":"Remove jobs corresponds to obselete featuresets\n\nAs we are removing the jobs which are using the obselete featuresets [1]\nso removing those jobs from templates/layouts.\nAnd this branch is also EOL for TripleO.\n\n[1]: https://review.opendev.org/q/topic:%22cleanup_featuresets%22+(status:open%20OR%20status:merged)\n\nChange-Id: Ifd1b570a7288c8c3ef322b6ad931d3ac9dcc354f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0449fb882a133c3b05264da4961c5dce26dcefcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0449fb882a133c3b05264da4961c5dce26dcefcc"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
