)]}'
{"id":"openstack%2Fheat~813155","triplet_id":"openstack%2Fheat~stable%2Ftrain~Ic0ecf8b241a87644d7200635fadc5795066b7b37","project":"openstack/heat","branch":"stable/train","topic":"volumev2-removal","hashtags":[],"change_id":"Ic0ecf8b241a87644d7200635fadc5795066b7b37","subject":"Partial Revert \"Change py2/grenade jobs to non-voting\"","status":"ABANDONED","created":"2021-10-08 09:26:34.000000000","updated":"2022-11-30 05:35:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11457ae11701ee6c145382f6c33bdb174a706dc1","_number":813155,"virtual_id_number":813155,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-08 14:22:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 11:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-08 14:34:07.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"}],"messages":[{"id":"c600c5f9eb912798cafca7a13c204b4b8b93b935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-10-08 09:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e27c961abea3f36f0994806b5b502bd9723e9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-10-08 09:27:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2ca3c91ee369be03542a19fdcfb960806618b6f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 11:00:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/217135d9dd974de7abeaf239430420f7 : SUCCESS in 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71b117d601034511a060f46f4f895389 : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa089c0ce704316935e7d4a79110361 : SUCCESS in 8m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/542ce41bfd394ab88f457be322e47c3d : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/936fe28ebe324e76a690aa9fac292ee8 : SUCCESS in 9m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb2aca06244b41bb9ddebe2e9e96625f : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8929743c996148c9b744cfccf89d4e66 : SUCCESS in 8m 43s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/f9a895509c7943878969a140baa01c56 : SUCCESS in 1h 16m 52s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0d50a02404fb41a2b027f0e4d9c67a57 : SUCCESS in 1h 26m 32s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/ca042f11ac9a4282a1ab17b9ccc058e1 : SUCCESS in 1h 16m 26s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/3430a8a1eea74496a9314a20c8588ef5 : SUCCESS in 1h 18m 01s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/fd89f63b0a50498984e47546733ce775 : SUCCESS in 1h 16m 08s\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/05a31e36249a4133b6471f17310dbd92 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f8c6848d10d6e216210a3020d0eaf6b0341ee834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-10-08 11:05:24.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"b6248bc0d526553f0a1832750c279abedd28fb94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 12:32:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad8f8c77cf0e4fd7a7ad6e0eed1ca42e : SUCCESS in 9m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/548cbe10f45b483fae5446e71259b585 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6f9c54636c4946a8991ff8e6742f78 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17f7ed7587604e71a17c3c666f8dbdcb : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a75a556351dc49feb9b320ee254f37ec : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ea838f94c7f48da93427be860b3c0d1 : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed4cd93ff3124b028319e6429bf0ecbc : SUCCESS in 7m 24s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/9c92320b3dd74fc78994001442f2e9ca : SUCCESS in 1h 19m 49s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/aeb10aa4b8644a83b12baa6997d02faf : FAILURE in 46m 12s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/bbc2f818a1a747a38aa18882ff2df51a : SUCCESS in 1h 11m 55s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/55a5e1958a964125be2d2ab6470b33fb : SUCCESS in 1h 21m 57s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/df27e62422c541dea91f45932dfcb8e0 : SUCCESS in 1h 14m 43s\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/b05113b1e0904c9e9ce11a99ff898c31 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":3},{"id":"d5bf8c9ba650231b7519a3690282b51f04763137","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-10-08 12:35:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"726a4c9f7dce5744dcccab9d17d13dad461cc81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 14:22:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3afacb90f30442380348634d691be0c : SUCCESS in 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6694379138794ba4bf94350db37e1122 : SUCCESS in 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb904283cca49f5a1d4269338924837 : SUCCESS in 7m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b454bae91784604b3663a523102865f : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af68f72569b0445ca62a09105b75bce4 : SUCCESS in 8m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7388c25f2334bde974fc98b87717f74 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63cba9ef5e3845e6a333dde2e252e6fd : SUCCESS in 8m 40s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/d8bca31423bc42cbafe7dbe0e18372e3 : SUCCESS in 1h 15m 10s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c230d5d852304b3aa285fb61dab64d75 : SUCCESS in 1h 38m 14s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/c310f708f47245e8b90b8d58d985452e : SUCCESS in 1h 03m 20s\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/6091e60d8a814d919355afc98cfef470 : SUCCESS in 1h 16m 43s\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/ea22c0412b794ccd95617e701e9abbbb : SUCCESS in 1h 23m 53s\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/1d84a534d6ff40478a02f760ad55dc41 : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":3},{"id":"11457ae11701ee6c145382f6c33bdb174a706dc1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-30 05:35:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18e0e661a0dc3c042bf7083c9b2b9805ec1e8fa0","revisions":{"002daba81dc73695fbcd2d087851ae364a4a5b41":{"kind":"REWORK","_number":1,"created":"2021-10-08 09:26:34.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/55/813155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/55/813155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/55/813155/1"}}},"commit":{"parents":[{"commit":"db86090ce775e9a07d8cf598df9bec90087f185f","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/db86090ce775e9a07d8cf598df9bec90087f185f"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2021-10-08 09:25:06.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2021-10-08 09:25:06.000000000","tz":330},"subject":"Revert \"Change py2/grenade jobs to non-voting\"","message":"Revert \"Change py2/grenade jobs to non-voting\"\n\nThis reverts commit ede47008bcfe3acef2050476b42ae6a3bedb7a67.\n\nReason for revert: Those seem to work now.\n\nChange-Id: Ic0ecf8b241a87644d7200635fadc5795066b7b37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/002daba81dc73695fbcd2d087851ae364a4a5b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/002daba81dc73695fbcd2d087851ae364a4a5b41"}]},"branch":"refs/heads/stable/train"},"44ba9c5a61538ea6dd0850a9e3eea84f5a909196":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-10-08 09:27:41.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/55/813155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/55/813155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/55/813155/2"}}},"commit":{"parents":[{"commit":"db86090ce775e9a07d8cf598df9bec90087f185f","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/db86090ce775e9a07d8cf598df9bec90087f185f"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2021-10-08 09:25:06.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2021-10-08 09:27:13.000000000","tz":330},"subject":"Partial Revert \"Change py2/grenade jobs to non-voting\"","message":"Partial Revert \"Change py2/grenade jobs to non-voting\"\n\nThis reverts commit ede47008bcfe3acef2050476b42ae6a3bedb7a67.\n\nReason for revert: Those seem to work now.\n\nChange-Id: Ic0ecf8b241a87644d7200635fadc5795066b7b37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/44ba9c5a61538ea6dd0850a9e3eea84f5a909196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/44ba9c5a61538ea6dd0850a9e3eea84f5a909196"}]},"branch":"refs/heads/stable/train"},"18e0e661a0dc3c042bf7083c9b2b9805ec1e8fa0":{"kind":"REWORK","_number":3,"created":"2021-10-08 11:05:24.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/55/813155/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/55/813155/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/813155/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/55/813155/3"}}},"commit":{"parents":[{"commit":"db86090ce775e9a07d8cf598df9bec90087f185f","subject":"Use Block Storage API v3 instead of API v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/db86090ce775e9a07d8cf598df9bec90087f185f"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2021-10-08 09:25:06.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2021-10-08 11:05:19.000000000","tz":0},"subject":"Partial Revert \"Change py2/grenade jobs to non-voting\"","message":"Partial Revert \"Change py2/grenade jobs to non-voting\"\n\nThis reverts commit ede47008bcfe3acef2050476b42ae6a3bedb7a67.\n\nReason for revert: Those seem to work now.\n\nChange-Id: Ic0ecf8b241a87644d7200635fadc5795066b7b37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/18e0e661a0dc3c042bf7083c9b2b9805ec1e8fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/18e0e661a0dc3c042bf7083c9b2b9805ec1e8fa0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
