)]}'
{"id":"openstack%2Fheat~816601","triplet_id":"openstack%2Fheat~stable%2Fxena~I96ab14871ee8c5d5b83cc0cd4abc840ef0218ca8","project":"openstack/heat","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2021-11-10 22:09:40.000000000","reason":"Ian Wienand replied on the change"}},"hashtags":[],"change_id":"I96ab14871ee8c5d5b83cc0cd4abc840ef0218ca8","subject":"Fallback to upstream for Fedora image","status":"MERGED","created":"2021-11-04 01:17:21.000000000","updated":"2021-11-11 01:34:21.000000000","submitted":"2021-11-11 01:32:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816601","meta_rev_id":"c06d1fa0325653fe3951aece9ac81ccf800ff1ee","_number":816601,"virtual_id_number":816601,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-11 01:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-11-10 05:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-11-08 10:27:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-11-10 05:03:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-04 01:34:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-08 10:27:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-10 05:03:07.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"d788577c55cf081a211706e64a0eb27056b80bc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-11-04 01:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0fbb33d4003e59256bf0db7befc416e573b5441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-11-04 01:19:14.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"13b8c8f3d37cb8969433f3f66af932e865d7b142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-11-04 01:29:53.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"04c8e79b2df5cbfa0ad1a9b6ee653f610cf2adcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 01:34:10.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"ac1160b2d07849adae1976079081a1b39074274b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 01:34:14.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"fb4466458d445dc99d866dc4a4ec7cd1b6c587b6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-11-04 01:37:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3984653ca65278a51b511c0312359b9e7f7da46f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 02:03:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ede32eda1d446f495d9fca649f63961 : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25ac76fab4c344bd8836220aa5d2a30b : SUCCESS in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a57cce39dd08a49cc61597e7f1d6c4411102be16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 03:25:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4433733a6b8e40ef9ca92aab2b04afc5 : SUCCESS in 13m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/727959c18a774c909d1fed1d54913664 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f37598e459430eb30a450634669807 : SUCCESS in 9m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/801363f1171347dcb66ba6a39a00c4e1 : SUCCESS in 11m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00aace3d514041e094d6c5a8e1028185 : SUCCESS in 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3611d5ccd47a41ae9e343a7627f3280b : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a077d6d251c498daacb188eb15a382f : SUCCESS in 18m 51s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1427b5e39a374451bda0a30afa3488c3 : SUCCESS in 1h 46m 03s\n- heat-functional https://zuul.opendev.org/t/openstack/build/66d5717909a143c3accde78c696c2f0b : SUCCESS in 1h 04m 28s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/213b49caa71c4e019c53886db1d217ec : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":3},{"id":"7e8b5fe1d8d980e41cdef17b69f0cfaad0c00fa9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-08 10:27:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c68b7cb256e1b245c6cd9fc0c318d0b77e22a8c0","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-11-10 05:03:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5bc3193c1443eb48ebd4c7a8cb22b66e4695d986","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 05:05:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d551099790a6c4f13d0c954694bb2c1b37062c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 15:46:41.000000000","message":"Patch Set 3:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"028015ac69e935f97bae70d7c7c1dd75479a4e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 17:54:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a642547254e437c960b7764c98bef05 : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/260970aa909e449ab6ea6867ab20f0a8 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/205df68f948146109acba963936d09fe : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a947f7d19420419793e9ea7c6062b8d8 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674ef07b5c634b8e89de74a699eda794 : SUCCESS in 15m 23s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/14a6f58b18d0432a9ede61169dea5251 : SUCCESS in 1h 37m 54s\n- heat-functional https://zuul.opendev.org/t/openstack/build/ceb117301c294e93891ed00dd22fdb5e : FAILURE in 1h 17m 00s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9f8a91f760fe42759238b37334009798 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":3},{"id":"330990cda097359be0afa8f18f33b637e86818d7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-11-10 22:09:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f1c14bd667d1f4be52c018a0b68bf0ce25530280","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 22:29:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/061faccd2a0d4bcc9b2ebf3ed41ed447 : SUCCESS in 14m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cea60e4b2b5641a9ae0da83f54753ead : SUCCESS in 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3af36f56ef23207df487bb61a1fac5df56edaafb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 23:48:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f616caa03b344bce82f7573439679511 : SUCCESS in 9m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/100d00bd3be14299867235113017e127 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2412d1675a0f4dd89d7ef701fdc0153c : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6009ba4954249c183dc09789532f117 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058e84839e654a4491ab4daa70270e55 : SUCCESS in 9m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da9d8d5b204b4f0bb44fa1902b5c0466 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3803be9ee9284f538109c97fa9014579 : SUCCESS in 12m 08s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/98edc27bc98b4b33ab8ccba2e858afc8 : SUCCESS in 1h 32m 38s\n- heat-functional https://zuul.opendev.org/t/openstack/build/e94e714a57684aaca647212afc2001c4 : SUCCESS in 1h 04m 11s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/0a864f6323534f67a333ac199e1faecc : SUCCESS in 1h 17m 19s","accounts_in_message":[],"_revision_number":3},{"id":"04578c7a5b53985a869a4eea938570806cc30985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 23:49:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b76a645850dafaf23e2fc8db88c300b682235286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 01:32:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/461b703b6aee4ee3b40eafc7068ec1ff : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6bad17aa4cb4a3e8f542defb7dc137e : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79de67c0ae4846438975e3673ea576c6 : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e25bb85d39d49029fae0db8f5a52924 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bb5773d0224bb18e5522bd70680f57 : SUCCESS in 13m 37s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c91dbb148b634f5bb471cf90b15be7d8 : SUCCESS in 1h 41m 37s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b99bbabfb82d4a6a83aaf96163e2db7a : SUCCESS in 1h 03m 58s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/2a8222302c2f43cca6f2cbaf53e4e72d : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":3},{"id":"2e8f4973f7a0d90d05118269b2fe5269898ad4f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 01:32:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c06d1fa0325653fe3951aece9ac81ccf800ff1ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 01:34:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10f237a0d8a4804a69b5dea0ff93e1c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"07ebedefb9a6e2cdedb29eef99496b1c0a4203b6","revisions":{"552eb1b561f287dbe6806c74d0fa8ab3dfe21e38":{"kind":"REWORK","_number":1,"created":"2021-11-04 01:17:21.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/01/816601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/01/816601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/01/816601/1"}}},"commit":{"parents":[{"commit":"74164424016702bac6d52c6618360287ce7ea461","subject":"Fix ServerGroup create","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/74164424016702bac6d52c6618360287ce7ea461"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-11-03 23:04:58.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-11-04 01:16:56.000000000","tz":660},"subject":"Fallback to upstream for Fedora image","message":"Fallback to upstream for Fedora image\n\nOpenDev infra only keep around the latest two Fedora releases in their\nmirrors.  Probe for the image from the local test mirror, but if not\nfound, fallback to upstream.  This will be much less reliable, but can\navoid gate breakage until new images can be used.\n\nCherry-Picked-From: f7574531d3dc3bbbf55fd7c17d55a10b5320c5ff\nChange-Id: I96ab14871ee8c5d5b83cc0cd4abc840ef0218ca8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/552eb1b561f287dbe6806c74d0fa8ab3dfe21e38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/552eb1b561f287dbe6806c74d0fa8ab3dfe21e38"}]},"branch":"refs/heads/stable/xena"},"60b250df3d6177b8bbf32e65f0097b94f87c2bc0":{"kind":"NO_CHANGE","_number":2,"created":"2021-11-04 01:19:14.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/01/816601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/01/816601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/01/816601/2"}}},"commit":{"parents":[{"commit":"74164424016702bac6d52c6618360287ce7ea461","subject":"Fix ServerGroup create","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/74164424016702bac6d52c6618360287ce7ea461"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-11-03 23:04:58.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-11-04 01:18:50.000000000","tz":660},"subject":"Fallback to upstream for Fedora image","message":"Fallback to upstream for Fedora image\n\nOpenDev infra only keep around the latest two Fedora releases in their\nmirrors.  Probe for the image from the local test mirror, but if not\nfound, fallback to upstream.  This will be much less reliable, but can\navoid gate breakage until new images can be used.\n\nCherry-Picked-From: f7574531d3dc3bbbf55fd7c17d55a10b5320c5ff\nChange-Id: I96ab14871ee8c5d5b83cc0cd4abc840ef0218ca8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/60b250df3d6177b8bbf32e65f0097b94f87c2bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/60b250df3d6177b8bbf32e65f0097b94f87c2bc0"}]},"branch":"refs/heads/stable/xena"},"07ebedefb9a6e2cdedb29eef99496b1c0a4203b6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-11-04 01:29:53.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/01/816601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/01/816601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/01/816601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/01/816601/3"}}},"commit":{"parents":[{"commit":"74164424016702bac6d52c6618360287ce7ea461","subject":"Fix ServerGroup create","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/74164424016702bac6d52c6618360287ce7ea461"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-11-03 23:04:58.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-11-04 01:29:53.000000000","tz":0},"subject":"Fallback to upstream for Fedora image","message":"Fallback to upstream for Fedora image\n\nOpenDev infra only keep around the latest two Fedora releases in their\nmirrors.  Probe for the image from the local test mirror, but if not\nfound, fallback to upstream.  This will be much less reliable, but can\navoid gate breakage until new images can be used.\n\nDepends-On: https://review.opendev.org/c/openstack/heat/+/816602\nCherry-Picked-From: f7574531d3dc3bbbf55fd7c17d55a10b5320c5ff\nChange-Id: I96ab14871ee8c5d5b83cc0cd4abc840ef0218ca8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/07ebedefb9a6e2cdedb29eef99496b1c0a4203b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/07ebedefb9a6e2cdedb29eef99496b1c0a4203b6"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
