)]}'
{"id":"openstack%2Fheat~817894","triplet_id":"openstack%2Fheat~stable%2Fwallaby~I6433bd4e7e378fefccedaf96035f847bfb1efc60","project":"openstack/heat","branch":"stable/wallaby","topic":"wallaby_testenv_uc","attention_set":{},"removed_from_attention_set":{"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2024-03-06 12:21:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6433bd4e7e378fefccedaf96035f847bfb1efc60","subject":"Install upper-constraints for testenv","status":"ABANDONED","created":"2021-11-15 04:46:13.000000000","updated":"2024-03-06 12:21:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"65061f294f6a8de1eba93dffb0675e193cae30a5","_number":817894,"virtual_id_number":817894,"owner":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-15 06:32:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-15 05:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-15 06:32:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 12:21:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"ccb57cea920aa214d380250df24fc4d69fd185e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2021-11-15 04:46:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9907a6615ea022322c5c425ff3ae1115823e4b97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 05:05:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5dbcaf702d3641f48fbf3103bd74594e : SUCCESS in 11m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0261531de44241dba57993598761581c : SUCCESS in 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa7473499940916511ac13e6c733a3a6ef8d18bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 06:32:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/007afab5d33f4cc99cde6a69ba5b9f54 : SUCCESS in 10m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6dbe0d1a01c45fb99f08576ea4d6b95 : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c10d25fe7d344003ada0e163cce152ff : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68ba0e8123824c6fa1a3f0b050074107 : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb06dd9c32cb4c6c86c3833269d21e11 : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d11d9f2d1964b85a07b21ffb761a0d0 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe406d6f6dc4809b4a77ce25ea2730c : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddb3b854d5e84b2994c0e553665f0cd5 : SUCCESS in 7m 59s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/479ed79a65124263a9fc10b4817ab283 : SUCCESS in 1h 39m 12s\n- heat-functional https://zuul.opendev.org/t/openstack/build/4f5ba6f0f31549aabded45520921dc9b : SUCCESS in 48m 34s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/72f9868852fb4fb7b481bbfb118aa782 : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"65061f294f6a8de1eba93dffb0675e193cae30a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-06 12:21:38.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/heat is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"124f765f83063430c50c2e214f11ffc95261e484","revisions":{"124f765f83063430c50c2e214f11ffc95261e484":{"kind":"REWORK","_number":1,"created":"2021-11-15 04:46:13.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/94/817894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/94/817894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/94/817894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/94/817894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/94/817894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/94/817894/1"}}},"commit":{"parents":[{"commit":"6211913213cdfb78e35a0e4096d51a7ef71c13f8","subject":"Fallback to upstream for Fedora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6211913213cdfb78e35a0e4096d51a7ef71c13f8"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2021-11-15 04:45:53.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2021-11-15 04:45:53.000000000","tz":600},"subject":"Install upper-constraints for testenv","message":"Install upper-constraints for testenv\n\nLatest sqlalchemy causes issues for us. We will\npip install upper-contraints in testenv to work\naround this issue.\n\nChange-Id: I6433bd4e7e378fefccedaf96035f847bfb1efc60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/124f765f83063430c50c2e214f11ffc95261e484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/124f765f83063430c50c2e214f11ffc95261e484"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
