)]}'
{"id":"openstack%2Fheat~818608","triplet_id":"openstack%2Fheat~stable%2Fwallaby~I3f132d4889723f7b4a3a416779ac5ee7663249b8","project":"openstack/heat","branch":"stable/wallaby","topic":"story/2009237","attention_set":{},"removed_from_attention_set":{"4257":{"account":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"last_update":"2021-11-29 15:47:19.000000000","reason":"Zane Bitter replied on the change"},"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2021-11-22 10:02:31.000000000","reason":"Rabi Mishra replied on the change"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2021-11-30 08:56:54.000000000","reason":"gibi replied on the change"}},"hashtags":[],"change_id":"I3f132d4889723f7b4a3a416779ac5ee7663249b8","subject":"Fix timeout calc of nested resource groups","status":"MERGED","created":"2021-11-19 17:21:51.000000000","updated":"2021-11-30 12:07:59.000000000","submitted":"2021-11-30 12:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818608-story/2009237","meta_rev_id":"c25741cbc2cd64d41a5cfa4862ae5a994ab8de8b","_number":818608,"virtual_id_number":818608,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-30 12:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-11-23 11:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2021-11-29 15:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-11-29 15:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 17:22:21.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2021-11-19 17:22:21.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-11-19 17:41:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-19 19:06:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e462f0a4bf0f1df1c4aa5d397ca9a74cf9df48d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-19 17:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8772bac0f356b20fdf03513e23c5622925bcfbce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 17:41:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5917a81c4c114e31a18475692569dcfc : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f316ef5d6f441229f2c0e0e6ac9fca8 : SUCCESS in 17m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37ba08f901c7746bb6cb4d921f70bb6abf581b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 19:06:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bdca7b3aa964f6a9ea1547b96d85019 : SUCCESS in 11m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a626ef4866d64210977a855f7686d3d1 : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dcddb901b1341f59d40682dbb3ac8a5 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93ae9d3b87e941ed8693573991c7904f : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2113ae4a0694419866774d25e555f38 : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b168f79c5e241628fd3c759579ec2d0 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/847fd9a034b747bbb6bc50825b840c19 : FAILURE in 8m 39s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/ae632645c52a4c14a549b61a6cdc8d1d : SUCCESS in 1h 36m 39s\n- heat-functional https://zuul.opendev.org/t/openstack/build/91fd2d3e485f4092a43b47ac0895c730 : SUCCESS in 1h 02m 13s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/c2486304979d4fdeb1f85b5117b71c71 : SUCCESS in 1h 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"63559dc2b9e2b7aa48cfa2d196ce1f440e97da10","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-11-22 10:02:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7478bd56658b2350ddbd53b6954c6592345b214","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 10:27:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2ecee826d1e49f693245aac69e21e8d : SUCCESS in 15m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89705d8e1adb4ae1a165ee5548f8ad88 : SUCCESS in 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eaca0bdb5fa2c7c3de8a15c584959346c7555e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 11:38:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d107cd63658f409389294987a113abd3 : SUCCESS in 8m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54771f0c5a4a4e3cbee1cbf662d28830 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934d8bea9fbb4e21978e7e5e95610904 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9379af72c7f4563bd40bdcd012585d0 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b53a53d5b6c7414f92b3e5b9b26dd229 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96df316e72c2458394c75dd1959a5766 : SUCCESS in 11m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e83b48a6eee417d8942994455528e49 : SUCCESS in 11m 36s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/ae0372c4738444c19c51c079c77f634a : FAILURE in 1h 32m 55s\n- heat-functional https://zuul.opendev.org/t/openstack/build/54d7ff51a3e449cf8cd475825e355dba : SUCCESS in 53m 06s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9ede87fd77cc4dd2b91b959578f8038e : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":1},{"id":"71e0c5eed31d72035ad70ced51821b45a23a2ef8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-22 13:06:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0134ba71009df45917a43c686a0ae4cbf9238e81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 13:30:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d106d0bcbcc4243bf6a13108cfe2769 : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8712d32e32742df9444eb7d90427f67 : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25223f743d1212a2748baa1536944137e2077c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 14:44:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e35376e1c87c44b89db5030cf618dd72 : SUCCESS in 10m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06b4dda1e7064cc1a5c5dd6fc39da6f1 : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec2b8ce006e427e87adf610c9d08a08 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3de89865b9d4fd48017ee0624102d5e : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd888c47cba445eea996550deebc151e : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5607febf274402998eeee4f6a5778dd : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8720da0b2cae4dc989ee136fe55dc278 : SUCCESS in 10m 44s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/547491a134284b029599bdc472674675 : SUCCESS in 1h 33m 59s\n- heat-functional https://zuul.opendev.org/t/openstack/build/bea1dd382feb481483ce2341962d1c8d : POST_FAILURE in 59m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9f8eb31074e34095b44e885fb761298c : SUCCESS in 49m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c376d601f6278da9c0498fffa8661333b0dac928","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-23 09:29:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d3dffd964440070fe105d17da7f8db9b3ae8647b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 09:50:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac14b820606a43c1aa1c17f136481047 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8fdd3bf48e74b3dad318e6311ad402e : SUCCESS in 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3938f0c4f123d136fc3843349573d36ad1d3d441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 11:09:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3dcfdb4a86d4042b0fec344b1b1438b : SUCCESS in 11m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9524713d18424600988d0ce9da4923a0 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f95593be0a444ff99f3cb734846c518 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb0d0a880cd4410186680be3b263e21c : SUCCESS in 11m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a68f504fa24746248173bad8cc51f89c : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48c5dd9cefa84a79a0075381b5659343 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25db1d67ab2f437db64535b9a7ab3761 : SUCCESS in 12m 35s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/81af841d16c742a6af518391d1584428 : SUCCESS in 1h 37m 49s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7bd106379e8045488e650e7e64a21884 : SUCCESS in 56m 25s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/811561c2c033492495649b06913de4ec : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a0f30bb2b614524c8f570218f3d00d86be42569a","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-11-23 11:41:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"781d24a6e1996b4f379f55c3fbf65d53ead63473","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2021-11-29 15:47:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01126ab206064a826919148a7768138b26b822f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 15:48:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e130c9ba3a441a5341c54f963e2a175e0a3b5741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 17:30:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/187779d5d37444be81093a379bc163a8 : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f737f9444bd4100b55d04ad23249b0b : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41a666c27f64400797df910fa9055e2e : SUCCESS in 8m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/501b555c4c9b4eb5bfb355b0dda0a270 : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d349c77c214f4f4f93ff781a81a3ccb5 : SUCCESS in 11m 16s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9f3f10e68a6f4a0197345fa730e45fdf : SUCCESS in 1h 33m 42s\n- heat-functional https://zuul.opendev.org/t/openstack/build/6696b0ee63e4454689d3b27dd0376795 : SUCCESS in 1h 05m 17s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/fd7d0cc885a74d61ba456313f7e74cff : FAILURE in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"39c3225ffbd030c64ad6598f5553b0e7919145f5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-30 08:56:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d988196d9959b0d6fb401bb33143698f655348fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 09:16:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6aa5677bc844405a83d1d83e0025685b : SUCCESS in 14m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd8a7e47a4f940cd810471ee2a132dfc : SUCCESS in 17m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65dcffbbc25e57585f64c9a43a9c873aef70add3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 10:30:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4769d6bf0a1428dbbc8365de79e3f89 : SUCCESS in 11m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ea28025937d470bad3ff8d7c6818c1d : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e711973a88b43719bd10f9209ccaef9 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8f5a97842c3442a962c2491a8dc73fc : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/125e78ed89924187aae6ac1745f9bfb5 : SUCCESS in 9m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90cd1262b0244073854a60e1a8487c29 : SUCCESS in 10m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc7ffde3fe245f48b31ea3afe84a18d : SUCCESS in 13m 43s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/f1fb1d867e1d4268a881ad24c8a48783 : SUCCESS in 1h 27m 49s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c7ab06b998db4ac2a78e26aa7c497117 : SUCCESS in 1h 07m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/6ebd3fe1a4694b24bda5884ba65414db : SUCCESS in 1h 09m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c40bc895cfc41017de279df41d705b8618b74277","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 10:31:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4277c95c265bbc644da6476e88fb232e026623c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 12:06:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69e75b23e3a84bc6aca5bef334a90b58 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86dcfef9ca54d5a8b1453e4d89a8780 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/464eac32c4eb4f9bafcde453e667b735 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc4f2de4376346f9adf2f0f453284db0 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18b1b1305e94002a066fc210b45e5e4 : SUCCESS in 11m 37s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/b4370040212243229f9c4816efb6d016 : SUCCESS in 1h 29m 13s\n- heat-functional https://zuul.opendev.org/t/openstack/build/e170e1b8321d41cd992db08043e0de31 : SUCCESS in 1h 09m 18s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/44f1c258e9fb43609f3f95ad8599890a : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ce36d85754f5af45d0a507b096dbe347044d7613","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 12:06:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c25741cbc2cd64d41a5cfa4862ae5a994ab8de8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 12:07:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b520998a5844e2f9fff80a7895a172e : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6002015fec472e0c90c91055f1963f323a1793e7","revisions":{"6002015fec472e0c90c91055f1963f323a1793e7":{"kind":"REWORK","_number":1,"created":"2021-11-19 17:21:51.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/08/818608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/08/818608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/08/818608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/08/818608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/08/818608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/08/818608/1"}}},"commit":{"parents":[{"commit":"a5e713b62235fd91e7edc89cc951b30dcb2e7c86","subject":"Reproduce bug story/2009237","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a5e713b62235fd91e7edc89cc951b30dcb2e7c86"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-09-20 18:54:42.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-11-19 17:21:14.000000000","tz":60},"subject":"Fix timeout calc of nested resource groups","message":"Fix timeout calc of nested resource groups\n\nThe timeout calculation of nested resource groups are using seconds but\nthen it passing the value in seconds to the stack resource that is\nexpected it in minutes. This leads to a rapidly increasing timeout value\nin each nesting levels.\n\nThis patch converts the seconds to minutes before passing it forward.\n\nStory: 2009237\nTask: 43372\nChange-Id: I3f132d4889723f7b4a3a416779ac5ee7663249b8\n(cherry picked from commit ef731bca05ea683c30448a2db397ae631c121d95)\n(cherry picked from commit f3c74a6632c5ced8c25f5811315e3d943be13c17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6002015fec472e0c90c91055f1963f323a1793e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6002015fec472e0c90c91055f1963f323a1793e7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
