)]}'
{"id":"openstack%2Fheat~818655","triplet_id":"openstack%2Fheat~stable%2Fvictoria~Ida63e1d865dc0b2ac6a13594455962f53068c931","project":"openstack/heat","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2021-12-14 03:57:20.000000000","reason":"Rabi Mishra replied on the change"},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2021-12-14 05:17:18.000000000","reason":"Change was submitted"},"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2021-12-14 05:17:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida63e1d865dc0b2ac6a13594455962f53068c931","subject":"Provide install_command to pip install with upper-constraints","status":"MERGED","created":"2021-11-19 23:59:44.000000000","updated":"2021-12-14 05:20:14.000000000","submitted":"2021-12-14 05:17:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"818655","meta_rev_id":"9d281d628ae65e2d39529de571e388c54eac4840","_number":818655,"virtual_id_number":818655,"owner":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-14 05:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-12-14 03:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"date":"2021-12-08 10:35:17.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-12-14 03:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2021-12-13 15:04:42.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"}],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-20 01:41:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-22 15:18:55.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-12-08 06:07:59.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2021-12-08 06:07:59.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"REVIEWER"},{"updated":"2021-12-14 03:57:20.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"7735030c5a95f6ee7682e74ef31f6ae7f5c2f812","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2021-11-19 23:59:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"465c291ffcf7826839d0e9413b92cd64949a192c","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2021-11-20 00:02:16.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d96b21e6f10b0dc56770498d9feff2026fc3915","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2021-11-20 00:03:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8260a776cba5febfff5e8f231dfbb3fe04ef49df","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2021-11-20 00:05:00.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"08c8f5df29a3cc64c5b51e009c757479f0a0845b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 01:41:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8141b76672d54a79ad1b4418d559ca48 : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4dd12e5e0ea4c5e85c0b204d2148394 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b407cfeb66674a1889813559ae10bfbd : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a62d061b87124d3d9d2be4e62cd0f054 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79f2b449266844e2ae6f6739cb757f26 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024b27ee17724e4983b945ec16f03834 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256a731df67e491e9df83b9171bc812d : SUCCESS in 7m 08s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/2c47200963d64991a04831c70861767e : SUCCESS in 1h 19m 09s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/03db050f9d8e461fb779f8f976f3a734 : SUCCESS in 1h 26m 44s\n- heat-functional https://zuul.opendev.org/t/openstack/build/49d4ba9496a845aa86a6c878062c0363 : SUCCESS in 1h 10m 14s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/06c140705c424cbcb87351de66d1448c : SUCCESS in 1h 01m 42s\n- heat-functional-non-apache https://zuul.opendev.org/t/openstack/build/e3a3b1cc084c4eedbbd37c6106099b0e : SUCCESS in 1h 05m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a79b0f1caed3d2830ff77a9be1c6275e66fed20e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-22 15:18:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"276a9a8b03749cb7ef619c8ac3a756ad2c9af4f5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-22 15:43:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ff653ae9589b8926529e9975f191f97d10ea7b2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-22 16:32:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae15cb07fadf96d8b795949972cb0235372e6048","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-12-08 06:07:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"343edd88aa922049e5dce0c4078a4d762f267da3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-12-08 10:35:17.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bdf4bb082b70e86dff07c170adc1d7cee4ea58e","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2021-12-13 15:04:42.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"60bd259465ad38f6f3bd7d031ae90bf80381e501","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-12-14 03:57:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7be8dfc79f9441191dc7d88ebbcae743cb08452a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 03:58:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c199f0dfb2f9bdf0db471b4635b9dad4205b8bed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 05:17:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/862f4913c4bb4bc1b5b027afed641d24 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/955e0ad3726448cba20bcbc8d1c220cc : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e04ad3558d434d7c97bbb260a24e7f91 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/999bc9e79bf84ff3a6fdf80604984353 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c354d2ec5c094e05bf170f4163f0faaf : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2b2f5c4b244063a4ba4e4d46af82ad : SUCCESS in 7m 06s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ffd07f0895ab47ce8474554f69e4f845 : SUCCESS in 1h 08m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9f853ce87d4841108e33cc345ec7b94e : SUCCESS in 1h 17m 58s\n- heat-functional https://zuul.opendev.org/t/openstack/build/337e2d7288024a01a3e82dc540276652 : SUCCESS in 56m 18s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7032fe099d524d36bf84d3f8b58b35d6 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":2},{"id":"4c761abed338308afef0013ec840d0fe890963f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 05:17:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9d281d628ae65e2d39529de571e388c54eac4840","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 05:20:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a42ae47d9e8244e1857cb836727fb6dc : SUCCESS in 2m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e490d41f3054366aacd49209eb3c9c5 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbe9be5e7625d374fc6cfbb63b7e3ef5335bf330","revisions":{"d71b1aedd8946e73ffee79b9797e5d60e4106dfb":{"kind":"REWORK","_number":1,"created":"2021-11-19 23:59:44.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/55/818655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/55/818655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/818655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/818655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/818655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/55/818655/1"}}},"commit":{"parents":[{"commit":"c062de4f24f1c29d8e7c7bb031a9c9672a183fc9","subject":"Fallback to upstream for Fedora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c062de4f24f1c29d8e7c7bb031a9c9672a183fc9"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2021-11-15 04:36:29.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2021-11-19 23:59:24.000000000","tz":600},"subject":"Provide install_command to pip install with upper-constraints","message":"Provide install_command to pip install with upper-constraints\n\nSQLAlchemy latest (1.4.27) installed with docs job breaks it.\n\nChange-Id: Ida63e1d865dc0b2ac6a13594455962f53068c931\n(cherry picked from commit 7f15b14d800654474ccab26af40384ab4515a657)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d71b1aedd8946e73ffee79b9797e5d60e4106dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d71b1aedd8946e73ffee79b9797e5d60e4106dfb"}]},"branch":"refs/heads/stable/victoria"},"cbe9be5e7625d374fc6cfbb63b7e3ef5335bf330":{"kind":"REWORK","_number":2,"created":"2021-11-20 00:03:36.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/55/818655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/55/818655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/818655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/818655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/55/818655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/55/818655/2"}}},"commit":{"parents":[{"commit":"c062de4f24f1c29d8e7c7bb031a9c9672a183fc9","subject":"Fallback to upstream for Fedora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c062de4f24f1c29d8e7c7bb031a9c9672a183fc9"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2021-11-15 04:36:29.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2021-11-20 00:03:17.000000000","tz":600},"subject":"Provide install_command to pip install with upper-constraints","message":"Provide install_command to pip install with upper-constraints\n\nSQLAlchemy latest (1.4.27) installed with docs job breaks it.\n\nChange-Id: Ida63e1d865dc0b2ac6a13594455962f53068c931\n(cherry picked from commit 7f15b14d800654474ccab26af40384ab4515a657)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/cbe9be5e7625d374fc6cfbb63b7e3ef5335bf330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/cbe9be5e7625d374fc6cfbb63b7e3ef5335bf330"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
