)]}'
{"id":"openstack%2Fheat~826224","triplet_id":"openstack%2Fheat~stable%2Ftrain~I1b9293041f2ad92eac0e9bc9646e7b2d7c6f7fd0","project":"openstack/heat","branch":"stable/train","topic":"story/2008632-stable/victoria-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-02-25 03:16:11.000000000","reason":"Rabi Mishra replied on the change"},"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-02-25 18:20:43.000000000","reason":"Harald Jensås replied on the change"},"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2022-02-24 00:04:13.000000000","reason":"Brendan Shephard replied on the change"}},"hashtags":[],"change_id":"I1b9293041f2ad92eac0e9bc9646e7b2d7c6f7fd0","subject":"Use neutron client for server.addresses","status":"MERGED","created":"2022-01-25 18:58:28.000000000","updated":"2022-02-25 21:47:10.000000000","submitted":"2022-02-25 21:45:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826224-story/2008632-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":826223,"cherry_pick_of_patch_set":1,"meta_rev_id":"321def4f449543fb7c60f5811221f59bcf89aba8","_number":826224,"virtual_id_number":826224,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-25 21:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"recommended":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2022-02-24 00:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":2,"date":"2022-02-22 22:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-02-25 03:16:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-25 20:23:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 22:00:23.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2022-02-22 22:03:25.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2022-02-24 00:04:13.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"38cb790abca6f57858306ccb80bfb1bca3c256f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-01-25 18:58:28.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"888b671ad38c36ad60a19cb3610b2ccc11df72e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 20:23:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72ef27a067664112a6a1a68195853110 : SUCCESS in 9m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dfdaac1cc164eedb4b25bae81d91737 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b6dae8da6247b18cd695edf7aab3b8 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b3f5a0fe1ac44263a40e8768ee97c34f : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba20926e518b420d9cee3fcfd0a2a47a : SUCCESS in 10m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ec197ec697549eeb8315372418bdc3b : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9317f8054846baa84cb5a66d4a2c59 : SUCCESS in 8m 00s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ac7ed704767147e2a729294a587aaf5f : FAILURE in 1h 10m 39s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/3af548fef98a461194d375402bdde978 : FAILURE in 1h 21m 49s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/643538975271417cab00d168ae963a33 : FAILURE in 40m 03s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/6d6cfd04cf714eaf8bfad220152e6107 : FAILURE in 32m 28s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/3b6d021582c14f2f90b849ddcc8bcc23 : FAILURE in 56m 30s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/3e3d9a685c8d471eb4d61145c4400f80 : FAILURE in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"129104f83ec06a9f3c4ef124eb216b72b76c72e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-12 05:09:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"84f3efe2592e94f3c8dc1963659b1ae4d267be11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 06:41:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66ea8a052d054db293c7c5141cf39ef1 : SUCCESS in 9m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74cfd1b21303446bb97c39b090381b83 : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03079be9922b4fd1883c4238dfce9980 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74ca3b721b134a628e4e6875903f7823 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65888692724b41b3b259bded642b1442 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23d827c6d84f41998c08690690476f9e : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fa025b24834f0487614d7c665f39ae : SUCCESS in 6m 58s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/78c4641202424394b6d3e4a6b2ba6b46 : FAILURE in 1h 11m 42s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/033676954c38400ca281c4769d90eb6c : FAILURE in 1h 23m 27s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/3ed91b4f10644f9b9843d7eb009532e2 : SUCCESS in 1h 17m 52s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/063c61282fb14e3b9659754d6fe7a79f : SUCCESS in 1h 25m 37s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/01e568b0e0cd436b8d50b3fd50b27ca3 : SUCCESS in 1h 21m 19s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/ccf51e5930b048debe9f48d7ed12f04a : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d5f26e187f7706f59826075ffbaaeeff7035a901","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-02-22 22:00:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d1339aa32f91bcc0b69ca7e7479b2fedae5e061","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-02-24 00:04:13.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c93bc8721ec00792d30b34d08710daf5420d366b","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-02-25 03:16:11.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2e55c8c6637f87ef683ecb4219f29d61f749eb65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 03:17:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"85fa295737a9982797a4a1b0955124f8bd73151f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 04:35:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/588f501d211b4c56a19cbb38ae015fce : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7efbbc520042359007367a5fd71d1e : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aaf3041fbd194c70a24aec2104e0d519 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/923cffeef6e44154b607831271d0388f : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48c4b127b7d04dc3a605c50186c8a4ce : FAILURE in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2377643ba7404aeaaf406df32eff8091 : SUCCESS in 7m 01s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/2372377cc39d4dec8a8e81f7aa93a5b3 : FAILURE in 1h 04m 35s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/98eaa50764664a9e8d588a47691e2954 : FAILURE in 1h 04m 09s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/2027edccfef14bffa3ac5ac85b001ad7 : SUCCESS in 1h 16m 43s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/90290c971c944e32800bb0badc6efb71 : SUCCESS in 1h 02m 31s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/cb2ad08b21fd4f638e0859c3304fdfb8 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":2},{"id":"a1fcfd69628fc1ccfa993ba8e175830dfe7fbd0a","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-02-25 04:44:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fd59fb552d536c51d4f1d916bbc2573a21111f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 05:44:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33b523582eb947c9afb89790f59d5e2a : SUCCESS in 8m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6591339caaf44a8e95eeaadc98ed4312 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f4177875c34cb299fde2c7855154a3 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/440eac51effe4a5d9d7684ffaaeadc49 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6206a01d05a441c99de10641091b7dd : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3ed17faf91746e1a613c658efb58505 : FAILURE in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7f92ffc8c0466f8b9c6db06629ecdc : SUCCESS in 5m 59s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/2377cc59f568478b89b295cbd68d1f5d : FAILURE in 47m 56s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c362a61aadaf4a4da7b10e75308de3d5 : FAILURE in 57m 05s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/e1a3b5584f9745eb878e27d4c86dd679 : SUCCESS in 57m 38s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/3a72c63996ff476bbcf67f2ea3ff7935 : SUCCESS in 56m 42s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/0198b266c9054047b591b78a8f3c57ae : SUCCESS in 58m 00s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/40fca98e88b34f08a0e1a41aafb60cb6 : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":2},{"id":"95f5b3dc67b35c8b0f9cb4d3e09e5af66ef92287","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-25 10:51:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d930c252d2a4eb1c833ecca7ef12c415a4532498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 12:16:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af0de7935bd84d78a14173043c8fc1f2 : SUCCESS in 10m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4996b2af2f5b4424a4ade221b4747bc2 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae4b0120c8014e17bdbe8e5a9888138e : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b00cfd73d7b4c7dbd9af26b05eaca71 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7ebc73be294d2f9994878e83740753 : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64ff0479ea284053a43fc95c70c48a43 : FAILURE in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d69efe32e24408a23946b0369ede57 : SUCCESS in 6m 42s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/6a2f6e1d08a74fd4b873bf45c177dc82 : FAILURE in 48m 57s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/435ddbb4d80a4daeb04eee4900e224f7 : FAILURE in 1h 05m 53s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/dbfe3767176448abb7c6d1afa6d950c3 : SUCCESS in 1h 09m 14s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/82c2c1b734e8413b910728838eb20559 : SUCCESS in 1h 06m 45s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/14142c1f39c64c60b9edb1cb70cf1a1b : SUCCESS in 1h 13m 24s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/417c27064fc34e94b200dba7bf3f6d86 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":2},{"id":"126ee178c1c8964c79c391cc7dd30f0475c6d643","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-25 12:53:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db8a5c5b0621ef01354be2abe99f8db8abdf207d","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-02-25 13:11:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"736b94b1140ef9e072e8a864979cdc18d7fccbfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 14:47:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59cef4bfd9aa47199d26666571691bce : SUCCESS in 10m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d8141df5a85a4decbd8d3924e6674ba3 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9960f90dfd0643658d6538b3f9860eb9 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0024c6c8659f4589b4be776e05f18957 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/932934045c9a4272a6d6fcc2947c9e67 : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33ec44e1453b4612a9f090826fdb56fd : FAILURE in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9fa58bb8e1543b0af86c9c8a4841139 : SUCCESS in 7m 17s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/70bbad5d60804b3db198d8a60cbf686c : FAILURE in 1h 07m 04s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/714a67da78b0461ab31e5d371427c0b3 : FAILURE in 46m 46s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/98fb74afc95a4990987ae747378da478 : SUCCESS in 1h 09m 42s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/a9e12272cb054d8d98d874464ed27d5e : FAILURE in 1h 19m 51s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/8af4c337dd204b689f1640717b7b5467 : FAILURE in 1h 33m 05s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/7f5215cf852d46c68f9f39b647ba0324 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":2},{"id":"4d0b3c37b2334835316a9a7db9149b6034727310","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-25 18:20:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"220cbe559ffe420b848bdda6a3e4cab72f3e7134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:03:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/723bc0a0f7bd4f40bd7f74f660002132 : SUCCESS in 11m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fe4cb01a88746868440f446b603c9bd : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa3c1dd1ae344e24861e43b43cfba56f : SUCCESS in 9m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ac320fa2ff84adc9de17d5a4ec9baf4 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98e8748f9a6943288c8ebda30c2ac714 : SUCCESS in 8m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e81c3a69b254dde9ee382a590c6bc04 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6769e5b5d34012bd1215932da410b8 : SUCCESS in 8m 16s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/0f80b42caf6b4fdaa9820bd74a4ca966 : FAILURE in 1h 10m 03s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/13d2b8b4194d43ad8d1c645d3d9e7195 : FAILURE in 1h 18m 53s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/8a4093616bc44e489b6d42bfcb3d3e1f : SUCCESS in 1h 18m 06s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/4ef4f03e69bf44d48d5a7a7817d179eb : SUCCESS in 1h 20m 01s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/5c461d51ca4d48d985e516ae14409491 : SUCCESS in 1h 19m 16s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/7a2d153a4a8f4be5aacee7cdbbe9aeea : SUCCESS in 1h 11m 30s","accounts_in_message":[],"_revision_number":2},{"id":"70cb2a35c339115c74ac2ded578ddcfb1d629f86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 20:06:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8e88d937113235ca72c474f9660ade63c4c9f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:45:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7d491d2a1a4420f8205dbe3ea82ecab : SUCCESS in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b87753d8c5a404b90467ee8ef116f4e : SUCCESS in 8m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a417eec445447f08c6e18d2cf4244b4 : SUCCESS in 10m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b9ef7433eb743f4a75e39442bb50516 : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94ce8b40a8bc410bad3c4d43fbf0626b : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c6a0d5919d43a982915cf65e062395 : SUCCESS in 8m 28s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/b0a9979e5d67489c9aeb1e7263f70fe7 : FAILURE in 1h 19m 42s (non-voting)\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1c87f2a11218461688b616b5a836ae72 : FAILURE in 1h 11m 54s (non-voting)\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/92db7db697c14af7ac4923395808fa7e : SUCCESS in 1h 07m 31s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/d1110a0bd6a24ea7a5c01ce572399ab1 : SUCCESS in 1h 31m 01s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/a268a224643a4f2f9da12c792e4b7a55 : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b25f4611e6b29f683e9324371e2fe7610860f401","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:45:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"321def4f449543fb7c60f5811221f59bcf89aba8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-25 21:47:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c71369e53747258316aead0bb9995d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09373756437234f9c79c12acc25229964bda61b0","revisions":{"2a8df779605f70f98987102c11d7662948c65569":{"kind":"REWORK","_number":1,"created":"2022-01-25 18:58:28.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/24/826224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/24/826224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/826224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/826224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/826224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/24/826224/1"}}},"commit":{"parents":[{"commit":"05708d9d959407a869eda1c5cb1583ffe1e5d7d1","subject":"Don\u0027t create deployment when changing transport","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/05708d9d959407a869eda1c5cb1583ffe1e5d7d1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-02-16 16:25:17.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-01-25 18:58:28.000000000","tz":0},"subject":"Use neutron client for server.addresses","message":"Use neutron client for server.addresses\n\nThe server.addresses (/servers/{server_id}/ips)\nendpoint can contain stale data causing attribute\nlookups to fail.\n\nThis change replaces the use of server.addresses\nand instead uses the neutron client to list ports\nwith \u0027device_id\u0027 matching the server id.\n\nStory: 2008632\nTask: 41843\nRelated: RHBZ#1902230\nChange-Id: I1b9293041f2ad92eac0e9bc9646e7b2d7c6f7fd0\n(cherry picked from commit 45750c603ac9965afd5b4e53ecff65977b887b95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2a8df779605f70f98987102c11d7662948c65569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2a8df779605f70f98987102c11d7662948c65569"}]},"branch":"refs/heads/stable/train"},"09373756437234f9c79c12acc25229964bda61b0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-12 05:09:57.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/24/826224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/24/826224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/826224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/826224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/24/826224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/24/826224/2"}}},"commit":{"parents":[{"commit":"bd38beda6608b5efa5330fd0619ec3487ecd8992","subject":"Fallback to upstream for Fedora image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/bd38beda6608b5efa5330fd0619ec3487ecd8992"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-02-16 16:25:17.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-12 05:09:57.000000000","tz":0},"subject":"Use neutron client for server.addresses","message":"Use neutron client for server.addresses\n\nThe server.addresses (/servers/{server_id}/ips)\nendpoint can contain stale data causing attribute\nlookups to fail.\n\nThis change replaces the use of server.addresses\nand instead uses the neutron client to list ports\nwith \u0027device_id\u0027 matching the server id.\n\nStory: 2008632\nTask: 41843\nRelated: RHBZ#1902230\nChange-Id: I1b9293041f2ad92eac0e9bc9646e7b2d7c6f7fd0\n(cherry picked from commit 45750c603ac9965afd5b4e53ecff65977b887b95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/09373756437234f9c79c12acc25229964bda61b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/09373756437234f9c79c12acc25229964bda61b0"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
