)]}'
{"id":"openstack%2Fheat~844247","triplet_id":"openstack%2Fheat~stable%2Fwallaby~I7d1017bcd7147329c4e187336daee1a1ae280fa9","project":"openstack/heat","branch":"stable/wallaby","topic":"f36-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-03 08:46:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d1017bcd7147329c4e187336daee1a1ae280fa9","subject":"Use f36 instead of f33 for testing","status":"MERGED","created":"2022-06-01 05:57:19.000000000","updated":"2022-06-03 08:48:33.000000000","submitted":"2022-06-03 08:46:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844247-f36-stable/wallaby","meta_rev_id":"d45f7d4806e5f7749a3d2304c657bb29189c773a","_number":844247,"virtual_id_number":844247,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 08:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-06-03 07:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2022-06-01 06:30:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2022-06-02 06:14:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-06-03 07:37:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 05:57:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2022-06-01 06:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 06:30:27.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2022-06-01 07:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 06:14:09.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2022-06-03 07:37:08.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"ee83a15816f470bc909a18a6ced839ef7a3ab22b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-01 05:57:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65fe820f90677429b6413e9bb2dc3880e99ebb6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 06:16:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce873df8f9aa4828916adf3ce27ff1ec : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d84f52013d614212890fe98a16dae098 : SUCCESS in 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02bf8467d17d30704f1de3d0f1a7b8320c89d406","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-06-01 06:30:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4309058510f1af33509fed9913a7c014d49a8720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 07:11:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/121fab2919a04884b1b89180c3b87950 : SUCCESS in 10m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e899da1485894a0c9bb9682255a11ae8 : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7183bf1853fb4cc6ad21d1723f485456 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b7c90fbbe744cab9f000889f712c335 : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/231807dfe03847f08502b3462c77cb0f : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87e98677080a468eab217dc834b2f8b9 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e8cb929fa5485b8be3383acb361c42 : SUCCESS in 12m 00s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/572888eff8194a5daca644fb376beeeb : FAILURE in 50m 46s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/2759e1fa5ba04ef78b187c6496bc565a : SUCCESS in 1h 07m 04s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/4b050155a71540e6baf2f0dd76b299be : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"08df2a2488fe5e7e6bd2e2b913922daf6cc840ea","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-06-02 06:14:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9a836bc91bd7f68d41504cef3f60471c4eb05b22","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-06-03 07:37:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1a311418a1220b4f2bc7e489aaddd41ca33d168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 07:38:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4389fe71f9fd002e7ce08f01eb0e1e769e6c2f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 08:46:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d2bbf7797104228b751ba7713878e16 : SUCCESS in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709db03fcd4c4f84aa28b07e1733aaef : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8394d91610940e2833bb3398c9b420e : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfe256045fab45b68f55719119994144 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d6d3eaf4834974b37405f78c80c20c : SUCCESS in 11m 26s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/f3f02848e6ea46d59526886269f1428f : FAILURE in 59m 53s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/a824c1befd8c4f4893dad0538679392b : SUCCESS in 1h 06m 39s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/1c2e58a9108d4e13b5bae07bcddfcda3 : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":1},{"id":"87d19171067dffe3d68ecb2be255e91ba852e4ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 08:46:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d45f7d4806e5f7749a3d2304c657bb29189c773a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 08:48:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0269c1a6d91d4b3dbd31336e2cbf269a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b20eeff4df8d8bcd6317f75a8e7eafed635ab6b","revisions":{"4b20eeff4df8d8bcd6317f75a8e7eafed635ab6b":{"kind":"REWORK","_number":1,"created":"2022-06-01 05:57:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/844247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/47/844247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/47/844247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/47/844247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/47/844247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/47/844247/1"}}},"commit":{"parents":[{"commit":"aa31864de4fe480674a0669c05a024ab28c3c429","subject":"Fix None comparision when sorting by `updated_at`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/aa31864de4fe480674a0669c05a024ab28c3c429"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2022-05-30 14:54:30.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-01 05:57:11.000000000","tz":540},"subject":"Use f36 instead of f33 for testing","message":"Use f36 instead of f33 for testing\n\nF33 is unsupported for a while. We should not use it for any testing.\n\nChange-Id: I7d1017bcd7147329c4e187336daee1a1ae280fa9\n(cherry picked from commit 78244c5f0f26064cd53b68e13fa8c4f0e235e698)\n(cherry picked from commit eefb921612b6d560e9f28767ba3a40e25bdd4f7d)\n(cherry picked from commit b32163d9e35e522577b99110ebd1614de55f5653)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/4b20eeff4df8d8bcd6317f75a8e7eafed635ab6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/4b20eeff4df8d8bcd6317f75a8e7eafed635ab6b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
