)]}'
{"id":"openstack%2Fheat~845409","triplet_id":"openstack%2Fheat~stable%2Fyoga~I6a8cffdc86c895eebe4269c5cd37841325566c54","project":"openstack/heat","branch":"stable/yoga","topic":"venv-tempest-stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-15 05:12:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a8cffdc86c895eebe4269c5cd37841325566c54","subject":"Use constraints when creating tempest venv","status":"MERGED","created":"2022-06-14 03:49:58.000000000","updated":"2022-06-15 05:13:53.000000000","submitted":"2022-06-15 05:12:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845409-venv-tempest-stable/yoga","cherry_pick_of_change":803890,"cherry_pick_of_patch_set":9,"meta_rev_id":"671982708b6b34d7efdea0b9abdbeb356f1a0dbc","_number":845409,"virtual_id_number":845409,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-15 05:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-06-15 03:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-06-15 03:57:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-14 03:49:58.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2022-06-14 04:04:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-14 05:12:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 03:57:00.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"7963a3b62de668b50c5addcc52b0f2ae459d7b43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-14 03:49:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ecb1a7e29c7d2949898093f42e4d871252052388","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 04:04:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15157054ce2d4e369699179e912d6827 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cea3a8dead7041cd95645aeac1ff4fd2 : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77623817aba99b63989d0fe9c4c959edb8f56504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 05:12:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/279ba18a0c1444f9844bc3154fc648ca : SUCCESS in 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c19bb7d302b842a7badb31504eda7cab : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2516889733499990fcfafdd4e8e216 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcac78d3840e4277b9ab37a2f0b85044 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ccd62548ed945918cfb9dfa45cac06d : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7383a914f2864d829da489804156776f : SUCCESS in 11m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/fa20d91f41174c68b55e09a5897f0f9c : SUCCESS in 1h 19m 56s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/9ecdbd4454ac492da8ba9b6b8140142b : SUCCESS in 57m 05s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/2b85b290b4f140b09110a18384c293f5 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d0092bc5b7aaf010457b98400c84b97ce530c91e","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-06-15 03:57:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ce3ae46090280866aeb2a1c2080b5f789def806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 03:57:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39fab9d0f71c1757f612140833527ad6a8949be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 05:12:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/244792600dfa4bf79f445793fd40b8e1 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21cf714b2884072a8fd5503293833ba : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc141fbee81542198acc561f3cec4ad6 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0345a54ff35941278b6078ac7db91bf9 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f00c0606a348fe89e6ce15a1467aea : SUCCESS in 11m 06s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1146bf99b956451b8fa4c505b8b392ef : SUCCESS in 1h 13m 39s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/35172b5c9162491e9fc4002834c04f33 : SUCCESS in 50m 48s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a403406e5f6545aa8f0ac3fda9d7f51a : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"427e41fd762ba3c3cff8e67d58725511e5ef3071","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 05:12:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"671982708b6b34d7efdea0b9abdbeb356f1a0dbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 05:13:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698a134595d24d0e82d49f3e3010828a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b95e6025f61417e7abc24a26d991bd26c57ee444","revisions":{"b95e6025f61417e7abc24a26d991bd26c57ee444":{"kind":"REWORK","_number":1,"created":"2022-06-14 03:49:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/845409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/09/845409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/845409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/845409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/845409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/09/845409/1"}}},"commit":{"parents":[{"commit":"eefb921612b6d560e9f28767ba3a40e25bdd4f7d","subject":"Use f36 instead of f33 for testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/eefb921612b6d560e9f28767ba3a40e25bdd4f7d"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-07-30 03:53:05.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-14 03:49:58.000000000","tz":0},"subject":"Use constraints when creating tempest venv","message":"Use constraints when creating tempest venv\n\nWhen running tempest we run in-tree tests as well. Lots of these\ntests including tempest tests may not work with new client\nlibraries. Let\u0027s use the relevant branch upper constraints when\nrunning functional tests.\n\nChange-Id: I6a8cffdc86c895eebe4269c5cd37841325566c54\n(cherry picked from commit 22c48f57bb29ed62a8c3249ab953043cfd2617d3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b95e6025f61417e7abc24a26d991bd26c57ee444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b95e6025f61417e7abc24a26d991bd26c57ee444"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
