)]}'
{"id":"openstack%2Fheat~845840","triplet_id":"openstack%2Fheat~stable%2Fxena~I2de594258231e3f1c217904afce88a83a8f94df3","project":"openstack/heat","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-20 13:29:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2de594258231e3f1c217904afce88a83a8f94df3","subject":"Make grenade job voting again","status":"MERGED","created":"2022-06-15 12:27:51.000000000","updated":"2022-06-20 13:32:04.000000000","submitted":"2022-06-20 13:29:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845840","cherry_pick_of_change":845839,"cherry_pick_of_patch_set":1,"meta_rev_id":"3fcd3dd820acaff28ac822fd0bfff6c65a8d486b","_number":845840,"virtual_id_number":845840,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-20 13:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-06-20 08:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-06-20 08:23:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-15 12:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 14:19:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 08:23:01.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"8d0cfcbcea271751a51303ad22c35e4ba33f8e49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-15 12:27:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"d4524ec0c0ddcc21ef3f641b9790da2e28241b85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 12:47:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcd4586ee7a54b10919e0dc5600e63aa : SUCCESS in 11m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea43df85f33049f1b72b568909341b4c : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f6148934b1e9f96ffc249375bf3f494e7baf928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:19:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26f84284d51a44939037108bdee2fe2f : SUCCESS in 10m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f379e03bd6641bfa343e2bbac7bebf1 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35826d4551e547e4a542a4587754d08a : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57dc8f18e8754ce5be5f18d800b88e90 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ef2fb00cd7a4c91bf6cf9c1a4786396 : SUCCESS in 8m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49054ae944fe421a8a0fbf9e7baa0d2b : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ab4a5d80694a4d8db2f13ab49ecf8d : SUCCESS in 13m 46s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/31efaccf23ae4f1f9a84f1e7bfdfaae1 : SUCCESS in 1h 44m 42s\n- heat-functional https://zuul.opendev.org/t/openstack/build/cbaf0d47003a4cc3a36a21184ee07983 : SUCCESS in 1h 06m 15s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/78e1012e094e4b42b34e9c9f86e3fd5e : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7ecdaa9ade242d0e3bfa8aa774fce7cc0e7c678f","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-06-20 08:23:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a80792e2fff7d8bb14fda3fe5757a40128e5d0af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 08:23:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a919c2bf1aa9b09b918ab79b749444b88afeb7ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 09:55:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8a13c908de43b391838c68d0b2ee95 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64983d70863149589e6c84fd37d3891d : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c975ab3765e4c288f56d2b804c951ca : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d7aab0324544c7ac41b3f7f67af243 : SUCCESS in 12m 10s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/7b7af5b8d2c44ed39f0d27434ab181b4 : SUCCESS in 1h 29m 51s\n- heat-functional https://zuul.opendev.org/t/openstack/build/0b19359a6a4744baacfbeb97f447c451 : FAILURE in 52m 08s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/8be2a2f07129447fb7a9a8872bc47d93 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":1},{"id":"80e553b40f801a4ee0805c283ec0523ffdba6396","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-06-20 10:03:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac2ce46cf88cd79b7bcad4ae044fb63e41d10c11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:24:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d92c5f957a7d49d08713a5eec3588501 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21704c9f18154ab686984ea67e5afda6 : SUCCESS in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"67ccd0e8f4619509a78edbaa6f2f6edd748aca7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 11:57:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5434d651f83f415686702ed83ef7c9f7 : SUCCESS in 9m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0628346ece194e0ab32bf8510a7fb8b5 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34ff23f332745d58323138ecec9c2b5 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b90b513329c247ee9e5b7583e117f1b9 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7feddc8418d44cca1493cc6ec806dea : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28776f08aae244978589b540180cf464 : SUCCESS in 13m 47s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/006acc9dc9f94263b13b27f614b21602 : SUCCESS in 1h 46m 26s\n- heat-functional https://zuul.opendev.org/t/openstack/build/2af36bf08b6647228cb40fcc4196e568 : SUCCESS in 1h 06m 45s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/87862e9fd45d4eb5a8c4c7e5112e12c3 : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"19bec8cd5865d7ad3e42a97b8616817d764e3b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 11:58:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75d84f55cd582d845e361b0fdbed86b1b8c4be56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 13:29:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db7a58e3b19403fb82a324563b00de9 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e7d06bdb1b64b26bc53de817c61192b : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52500ffb6d794776b885779c12211e54 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a31ccbca588a41a0a4067fb972b06a82 : SUCCESS in 15m 14s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9a7533ae983043d7b9748ce96ca8891d : SUCCESS in 1h 14m 34s\n- heat-functional https://zuul.opendev.org/t/openstack/build/fba2ac820b504a8a995a32febcb2a8ba : SUCCESS in 51m 56s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/f275089a34324eb2ae250ececbd28c3c : SUCCESS in 42m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4a3f1ace0ad0d02eccc66bfa527063d8e60d197e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 13:29:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3fcd3dd820acaff28ac822fd0bfff6c65a8d486b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 13:32:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/319a2f50b8e7498ba4a6a99baac68305 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3ec8946b3de50c76b2c3b08c2f7356842d3f45a","revisions":{"e3ec8946b3de50c76b2c3b08c2f7356842d3f45a":{"kind":"REWORK","_number":1,"created":"2022-06-15 12:27:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/845840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/40/845840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/40/845840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/40/845840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/40/845840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/40/845840/1"}}},"commit":{"parents":[{"commit":"24ea7c03972647ced60eb39c704d47e856bcf8fe","subject":"Merge \"Use constraints when creating tempest venv\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/24ea7c03972647ced60eb39c704d47e856bcf8fe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-03 07:35:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-15 12:27:51.000000000","tz":0},"subject":"Make grenade job voting again","message":"Make grenade job voting again\n\nBecause change I7d1017bcd7147329c4e187336daee1a1ae280fa9 was backported\nto the previous stable branch.\n\nChange-Id: I2de594258231e3f1c217904afce88a83a8f94df3\n(cherry picked from commit fee8c40c23fb160a4fe64e95a71f827528287690)\n(cherry picked from commit 03e4f9af62ed66a4df72f5da793f8ce4caf8ac3e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e3ec8946b3de50c76b2c3b08c2f7356842d3f45a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e3ec8946b3de50c76b2c3b08c2f7356842d3f45a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[]}
