)]}'
{"id":"openstack%2Fheat~848596","triplet_id":"openstack%2Fheat~stable%2Fxena~Ic6125c6af60c1007243523cc8510a17f49c7b5af","project":"openstack/heat","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-04 15:58:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6125c6af60c1007243523cc8510a17f49c7b5af","subject":"Fix parameter merging with merge strategies","status":"MERGED","created":"2022-07-04 12:32:15.000000000","updated":"2022-07-04 15:59:23.000000000","submitted":"2022-07-04 15:58:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848596","cherry_pick_of_change":848590,"cherry_pick_of_patch_set":1,"meta_rev_id":"1a776d24b5b5ad522bb5e21b48883e78659a07f3","_number":848596,"virtual_id_number":848596,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 15:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-07-04 12:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-07-04 12:51:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 12:32:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2022-07-04 12:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-04 12:51:09.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2022-07-04 14:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf64351c0d0d3726877f0086fce805b56144ccfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-04 12:32:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"02b5f695929fb5b8790c01c52898cd7dd79809e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 12:49:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae8ba83edb14f428bfa95feda99d3b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c00aee14e984d379e0e908aec9e9ae5 : SUCCESS in 15m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fee5570f87f44a51a51b4de356636fec : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06309958ad94cbf2dabc5cca8cb768a76b8cf026","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-07-04 12:51:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2aa7a97ef8b40c06d2a7a2332b83e42f054ca941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 14:13:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca8b914641e746ee84184c955b93e588\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f8c216c01414b05978f99df8a7e9309 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ce9feb04034a199e9eed1cfd9374c8 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7f21b8dec4e416b86787f5f8ed12bcd : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab2e7a92bcd842fbb6b64f34f01e55f1 : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fda61702c14a4d69ac63517af6681bd6 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73280f3715f417485304e5ad43d9214 : SUCCESS in 13m 04s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/318b7dc893994b5f9a4cd87676e6c7a0 : SUCCESS in 1h 38m 42s\n- heat-functional https://zuul.opendev.org/t/openstack/build/94ecde345db74da68c661e3f9da33a9b : SUCCESS in 57m 07s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/0d2e430ad8d14dafaa13f1db58522da3 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c6442e187da001bf50ffda0d22fe56ce30377024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 14:14:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5ed5a3ee58e906a698f1714870e67c100226c72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:58:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3a88f832427484d825a8fe1c06f6fca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85fe2826545a441e96f20ac8cd2855ce : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2346bc876cb04f43823a59d2bd70e863 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f67573cdea14f768a7b0e394706bffa : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b28162ebbd942f696638ffa5dae4085 : SUCCESS in 11m 18s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/bd07c281fc2d41ab93efa751f2685afb : SUCCESS in 1h 39m 25s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d3d3515e92f041c7b3dd0c3a0437b385 : SUCCESS in 1h 06m 14s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a8d3f140a31c48e2b6464cd0e4e4dadf : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":1},{"id":"025f4edebb65de08a118258e6b5f15ee9532d78d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:58:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a776d24b5b5ad522bb5e21b48883e78659a07f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:59:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/488589d8475447a1853548ccc444ede7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1e615c7a8544ff82f4d5790220966f : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8d09f347c2d478e7d69c940152a4d85fbc58f7c","revisions":{"b8d09f347c2d478e7d69c940152a4d85fbc58f7c":{"kind":"REWORK","_number":1,"created":"2022-07-04 12:32:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/96/848596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/96/848596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/96/848596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/96/848596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/96/848596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/96/848596/1"}}},"commit":{"parents":[{"commit":"09441e4d4cf72e3bfe4dcf7f11d30a0c4e631614","subject":"Merge \"Make grenade job voting again\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/09441e4d4cf72e3bfe4dcf7f11d30a0c4e631614"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2022-06-30 20:52:46.000000000","tz":330},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-04 12:32:15.000000000","tz":0},"subject":"Fix parameter merging with merge strategies","message":"Fix parameter merging with merge strategies\n\nEarlier users had to specify merge strategy for\na parameter/parameter_default in every environment\nfile where they are used(or the defaults for that\nenvironment was used). However, the strategy\nfor a parameter in the first environment was\nignored as there is nothing to merge.\n\nWith Iec3367667248d0b46196a6d569f3cbb3b200d842, we\nchanged to use existing merge strategies, but the\nstrategies in the first environment were still\nignored. This patch fixes it.\n\nTask: 45578\nChange-Id: Ic6125c6af60c1007243523cc8510a17f49c7b5af\n(cherry picked from commit d9fe7301b3af2db9fa56d59eb405add94730ba50)\n(cherry picked from commit 154691d21d24351edbe066fd23c731afad456b85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b8d09f347c2d478e7d69c940152a4d85fbc58f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b8d09f347c2d478e7d69c940152a4d85fbc58f7c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[]}
