)]}'
{"id":"openstack%2Fheat~848945","triplet_id":"openstack%2Fheat~stable%2Ftrain~I2de594258231e3f1c217904afce88a83a8f94df3","project":"openstack/heat","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-07-29 12:18:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8833\u003e replied on the change","reason_account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-29 16:18:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2de594258231e3f1c217904afce88a83a8f94df3","subject":"Make grenade job voting again","status":"MERGED","created":"2022-07-07 11:23:16.000000000","updated":"2022-07-29 16:20:51.000000000","submitted":"2022-07-29 16:18:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848945","meta_rev_id":"5cda0ccbe85fd9013abc0eee747eecf3bc7a0d17","_number":848945,"virtual_id_number":848945,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 16:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-07-29 14:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-07-29 14:39:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-07 11:23:16.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-07-07 12:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de4d7e2af4d5f23797e8dac8acdc2e452add7c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-07-07 11:23:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0baeedc14399c7029ef86bd32a0c217831175ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 12:36:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33f857787f7445c984967d2767d3bf5b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/904f8c23418546d18f8d06fa3fe35f64 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0561ed1c751946dbb49d48e5104bd667 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75d3a633eabe4dcdb1fdd8285a91cefa : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3da2d030b65f4dc6b74d6ed3dff752d4 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c00aa57c7cb1482aae2687eafb0ab7ea : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ae31491ae324ad6ab745a31d3f5513b : SUCCESS in 6m 51s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/a4704a2dbb684ba892162dbd26042388 : FAILURE in 1h 09m 32s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/12d1109a9e6c48658e081d64421c91dc : FAILURE in 34m 06s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/3c6c64f0d1274af5a57def16231bbffe : SUCCESS in 1h 10m 37s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/cb371207c705442ba1af577597cbcf78 : SUCCESS in 1h 09m 41s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/bd52a425dc464743899dbe75257ab93e : FAILURE in 1h 00m 22s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/c35b823c5fde4a288cfdcdc1b497da77 : FAILURE in 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"58fd5cd07aa24930acfa22bcf5bc32cc89b65cb9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-15 05:05:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9fd4241c9fcf6f81a66d0b9e85a54dafa3077730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:20:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c424e9a0eab47a4b711cc15aaf13686\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd6e00067d4141a796d58257f2cb97a4 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20956d69706142af84678eb5373d48a3 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf5b2e35204f484094a7db0b6533e340 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab56de1d7b0549b5bb7bc22029db2a89 : SUCCESS in 7m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/545145dd8b5d4530a61f63ca159773a4 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2346e0deb0294bbabdeead5a549f77eb : SUCCESS in 7m 53s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/f88c980a07064802bf9c9f4ab95ae8a5 : TIMED_OUT in 3h 08m 58s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/fa0fa68762744397a7f4fd3f2869f4bb : FAILURE in 1h 13m 14s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/7d85ca5a999146d9998f9e3dad93ff49 : SUCCESS in 54m 09s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/f1f36639789f4b9d9c1bd191a2bd3bee : SUCCESS in 1h 10m 41s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/e8db7e4c12954544b13d8bc0029a1144 : SUCCESS in 1h 21m 05s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/fc2455b9358b4b6f96769bb92f2774df : FAILURE in 18m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1e8dc7527b70f46266e5188737f0b6e0f8a74add","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-29 09:52:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"facd6610d5751e3b65e3ac232278ee9f8eabb9c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-29 09:52:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d23ebbbde597e0ee74f8d0daab91ce8b8c74e527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 11:25:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ce67c90d3bd430f9e38c58787c9e679\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35835e167fc84447a3dd1f5faa11a708 : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a00dc5144dab4b5e953eda2cdfb14bdd : SUCCESS in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f26aa75bfc39463fb6f01737b9aaa8d7 : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6afd7b259cbf4bf4bebb3153ee4e0db5 : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f84e8dcd048a45ee8f22d39213f30eb6 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab667e9164b4283b897f76e4ae85fe0 : SUCCESS in 7m 59s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/0ff0f7765dfc4df0ba0b3e765b6b112f : SUCCESS in 1h 20m 48s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/8b445e699f2144f4b08381edd28d756a : FAILURE in 1h 27m 43s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/bd0b92cc9ee54c65a65965ec00e0bc4f : SUCCESS in 1h 19m 46s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/8c5df0be7ec247e2a149324759193b86 : FAILURE in 1h 19m 22s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/723529af60034a88993d93306762ad75 : SUCCESS in 1h 20m 35s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/9901c398f83849dcb4cd89e444b63850 : SUCCESS in 1h 18m 30s","accounts_in_message":[],"_revision_number":2},{"id":"44367fdeb89600786f3f2c398f08e70088499139","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-07-29 12:18:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1672db6ec1118f2d6a867ef1699c0c9aa49c1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:50:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84682fd83bb4d5985a97023922ec15b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9b1dfee805445c98d3e480525a7fc90 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343684f041084ded8b2bfd8a887fd96d : SUCCESS in 9m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16be84ac1fc543399a0dd5d72f22b25e : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ebbe2f92bb04b00ba4b980ef054a513 : SUCCESS in 7m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae6ce4ac62e54ce1a9247f1521fe9580 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a3169d973e408fa74229e98d39ff45 : SUCCESS in 7m 26s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/cc8abeda83ff4ceaaba202f9095df18f : SUCCESS in 1h 22m 51s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/38a8793f64aa4a8f8a49da518e0994a8 : SUCCESS in 1h 10m 02s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/8df60c2f5f3d4e9fb60241575dd9aa0f : SUCCESS in 1h 01m 54s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/3d9330c49fcd45f6856f839084cb92cb : FAILURE in 57m 30s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/8911f6c56cf746f08c8f352c2fd8d11c : SUCCESS in 1h 04m 05s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/d8ad577dcc074dc1ae94f4124318e622 : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6d36f36ff757670df99134dd75d96a9f57fa80d0","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-07-29 14:39:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"71a02d61bac1f735b1fbf68b81a8a039550f0685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:40:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc1e32240678b93110663e0376a7ccecd322cc84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 16:18:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/356dae55dfdd418eae7843051f7188ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab44ccdeefb44ae8db088d003401cce : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e4d80284968431fbe9f0e64096f7ca2 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/718637102b604f759c8e1af4e244fb7d : SUCCESS in 8m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e0dfd02f1924e1dbb66266a8e2a1579 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e672efa60f9e493a9623057d579670b8 : SUCCESS in 8m 22s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/ba9ce2ec270c4fff89733b2f08d30259 : SUCCESS in 1h 21m 34s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4cad775271b146668889f9f531114ac7 : SUCCESS in 1h 31m 00s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/6834a6ae222a4c2789e7c3c21d9b43a5 : SUCCESS in 1h 31m 02s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/72b05de23eba443e96705017d40cf45b : SUCCESS in 1h 29m 03s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/3aab1f527e86411f9d56e85040b0bc36 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2b2bce5502006f627720ca7ac56008affaaceadd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 16:18:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5cda0ccbe85fd9013abc0eee747eecf3bc7a0d17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 16:20:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc1da2393fe44e8e940bd9ade0c703cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c72c456b2a440cbe8c6f770463e179 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94cc394fe4311f462dcbab24e68761644e60d01b","revisions":{"a084d20741367696120ef6e6cb7e80b2e0069041":{"kind":"REWORK","_number":1,"created":"2022-07-07 11:23:16.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/45/848945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/45/848945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/848945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/848945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/848945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/45/848945/1"}}},"commit":{"parents":[{"commit":"fc762f2304a5f75f34df95fb01153f0da863157f","subject":"Merge \"Reduce Heat engine workers\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/fc762f2304a5f75f34df95fb01153f0da863157f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-03 07:35:56.000000000","tz":540},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2022-07-07 11:21:52.000000000","tz":330},"subject":"Make grenade job voting again","message":"Make grenade job voting again\n\nBecause change I7d1017bcd7147329c4e187336daee1a1ae280fa9 was backported\nto the previous stable branch.\n\nBackport note:\nThis includes the following follow-up commit.\n\n(commit fee8c40c23fb160a4fe64e95a71f827528287690)\nUse constraints when creating venv for grenade\n\nAs a followup for change I6a8cffdc86c895eebe4269c5cd37841325566c54\nlet\u0027s use branch specific upper constraints when running tests in\ngrenade.\n\nConflicts:\n\tdevstack/upgrade/resources.sh\n\nChange-Id: I2de594258231e3f1c217904afce88a83a8f94df3\n(cherry picked from commit fee8c40c23fb160a4fe64e95a71f827528287690)\n(cherry picked from commit 03e4f9af62ed66a4df72f5da793f8ce4caf8ac3e)\n(cherry picked from commit e3ec8946b3de50c76b2c3b08c2f7356842d3f45a)\n(cherry picked from commit b4d159b87e606024d65bb285576c64bd1c72fecd)\n(cherry picked from commit d7582f0032e753fdee9f5b12f17d66fd89aa4e85)\n(cherry picked from commit 72715a1f2f32b3ff377180d3b7700b97ad46b16d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a084d20741367696120ef6e6cb7e80b2e0069041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a084d20741367696120ef6e6cb7e80b2e0069041"}]},"branch":"refs/heads/stable/train"},"94cc394fe4311f462dcbab24e68761644e60d01b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-29 09:52:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/848945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/45/848945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/848945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/848945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/848945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/45/848945/2"}}},"commit":{"parents":[{"commit":"3914946580f03710524ee0465144c354a1273dfe","subject":"Use openstack commands in upgrade plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/3914946580f03710524ee0465144c354a1273dfe"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-03 07:35:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-29 09:52:28.000000000","tz":0},"subject":"Make grenade job voting again","message":"Make grenade job voting again\n\nBecause change I7d1017bcd7147329c4e187336daee1a1ae280fa9 was backported\nto the previous stable branch.\n\nBackport note:\nThis includes the following follow-up commit.\n\n(commit fee8c40c23fb160a4fe64e95a71f827528287690)\nUse constraints when creating venv for grenade\n\nAs a followup for change I6a8cffdc86c895eebe4269c5cd37841325566c54\nlet\u0027s use branch specific upper constraints when running tests in\ngrenade.\n\nConflicts:\n\tdevstack/upgrade/resources.sh\n\nChange-Id: I2de594258231e3f1c217904afce88a83a8f94df3\n(cherry picked from commit fee8c40c23fb160a4fe64e95a71f827528287690)\n(cherry picked from commit 03e4f9af62ed66a4df72f5da793f8ce4caf8ac3e)\n(cherry picked from commit e3ec8946b3de50c76b2c3b08c2f7356842d3f45a)\n(cherry picked from commit b4d159b87e606024d65bb285576c64bd1c72fecd)\n(cherry picked from commit d7582f0032e753fdee9f5b12f17d66fd89aa4e85)\n(cherry picked from commit 72715a1f2f32b3ff377180d3b7700b97ad46b16d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/94cc394fe4311f462dcbab24e68761644e60d01b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/94cc394fe4311f462dcbab24e68761644e60d01b"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[]}
