)]}'
{"id":"openstack%2Fheat~849935","triplet_id":"openstack%2Fheat~stable%2Ftrain~I64d727daabf5755f92987e4da33c6663460b6d5d","project":"openstack/heat","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-29 09:52:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I64d727daabf5755f92987e4da33c6663460b6d5d","subject":"Testing requirements fix","status":"ABANDONED","created":"2022-07-15 04:14:06.000000000","updated":"2022-07-29 09:52:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c72c65c6a0ad3f3be9a3e30cc344dd12ba9abc27","_number":849935,"virtual_id_number":849935,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-15 07:26:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 07:26:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9af2c21b6470641b1bc56b605b9531bb1e29e929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-15 04:14:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a9ae1d70b7f93aa6aefa5042edd5afab3c40027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 07:26:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a9c519115ba4558b05171aa8ca33169\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4934a50adcd746a1868932de29a8d65f : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ccab05c5bd84ed49f867bd4984967b2 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5449d6ce76864389bb6ffc886b48e0e5 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22119511f80a4b45b9579ed137cfeb4e : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/824e191e8bc74234b2f118360d57a99c : SUCCESS in 8m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f32f4d25a904cc99b579c5ce3ecd143 : SUCCESS in 7m 19s\n- grenade-heat https://zuul.opendev.org/t/openstack/build/faea9c16c7104cc5a23957d46d8fd11b : TIMED_OUT in 3h 08m 53s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/8ffdf385faf54ddc9d65fc04ef473437 : FAILURE in 1h 08m 05s\n- heat-functional-orig-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/7fa7d28fd0584872bdd4679acbcf39ce : SUCCESS in 1h 34m 57s (non-voting)\n- heat-functional-convg-mysql-lbaasv2 https://zuul.opendev.org/t/openstack/build/a88bba69e081421096ed0e81069e608c : SUCCESS in 1h 04m 08s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-non-apache https://zuul.opendev.org/t/openstack/build/31ec038b7dcb4bf69072d91f9e105b0e : SUCCESS in 56m 10s (non-voting)\n- heat-functional-convg-mysql-lbaasv2-py3 https://zuul.opendev.org/t/openstack/build/15f7afb5754c43d2b867fcd2de3ed0e7 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c72c65c6a0ad3f3be9a3e30cc344dd12ba9abc27","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-29 09:52:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa020d5617e37e73872dc2932a4b06033c4b67c3","revisions":{"fa020d5617e37e73872dc2932a4b06033c4b67c3":{"kind":"REWORK","_number":1,"created":"2022-07-15 04:14:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/849935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/35/849935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/849935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/849935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/849935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/35/849935/1"}}},"commit":{"parents":[{"commit":"a084d20741367696120ef6e6cb7e80b2e0069041","subject":"Make grenade job voting again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a084d20741367696120ef6e6cb7e80b2e0069041"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-15 04:13:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-15 04:13:46.000000000","tz":540},"subject":"Testing requirements fix","message":"Testing requirements fix\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/849767\nChange-Id: I64d727daabf5755f92987e4da33c6663460b6d5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/fa020d5617e37e73872dc2932a4b06033c4b67c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/fa020d5617e37e73872dc2932a4b06033c4b67c3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
