)]}'
{"id":"openstack%2Fheat~853335","triplet_id":"openstack%2Fheat~stable%2Fwallaby~Ieba5f353b0fe8a9b187bac1f6ff1b3a8f3b896b3","project":"openstack/heat","branch":"stable/wallaby","topic":"story/2010095-stable/wallaby","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-09-01 08:13:51.000000000","reason":"\u003cGERRIT_ACCOUNT_8833\u003e replied on the change","reason_account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-01 10:03:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieba5f353b0fe8a9b187bac1f6ff1b3a8f3b896b3","subject":"Fix AttributeError in Python 3.8","status":"MERGED","created":"2022-08-17 06:22:20.000000000","updated":"2022-09-01 10:05:02.000000000","submitted":"2022-09-01 10:03:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853335-story/2010095-stable/wallaby","cherry_pick_of_change":848127,"cherry_pick_of_patch_set":1,"meta_rev_id":"ced9118fc3c2e0dcdd5aa9e3d0be0a980f95e32c","_number":853335,"virtual_id_number":853335,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-01 10:03:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-09-01 08:13:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2022-08-18 10:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-09-01 08:13:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 06:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-17 07:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-18 10:53:35.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2022-08-29 16:26:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"ce81d04e5b474dd361e1e330f467e449771a1b83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 06:22:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"63d4f26abec2a005b53194a6d931cdcb298e7769","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 06:22:51.000000000","message":"Topic story/2010095-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"8ceab23d0582779d9b16f71f7393809d96458ed7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 06:22:56.000000000","message":"Topic set to story/2010095-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"214da17e0110cdafbe2cb66e4c87dd1221c2c2ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 06:38:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4c1d6177224258be52b71d9f23d53c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1860ed77df114060a43924f75f9414fc : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a40c7443ce4c41009b7c4cd37c671ea6 : SUCCESS in 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3a7670ec98c5bf93ad67ddc2a067c968252492e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 07:45:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ca234ec7be4511bbed161ff9186af5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b3bfec1704b425393f067505ad39add : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723fbb58119e4348b3c5ea6fb0fe440d : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a94a602e04ee458eb63fb19783333395 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce8c7567ff194a20a5d289f501a86036 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bbc643f00d440e1858d5b8f49368afb : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a1888dbe4b94495b4d4a24d74bba8cf : SUCCESS in 9m 30s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/aac56ffa12094833b658fceefde216a7 : SUCCESS in 1h 12m 45s\n- heat-functional https://zuul.opendev.org/t/openstack/build/9e289214673b4ffd9182a3f321d0ff0d : SUCCESS in 49m 06s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ab1afd4f560647d29b96669c55232c49 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1813fffa20a9b2e6e471a18c47cd5941a49c1d61","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2022-08-18 10:53:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"75cb38e1e991a59b5d54f285b82877f56b7d043c","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-09-01 08:13:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe25ec5b43a7f7bcf72ecfa607aa608c20489a01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:14:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20914ffee293b232809b26876288704110635db1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 10:03:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47646cd328884b13811fa304f250c42e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a38e2201e9e429caa79c721c179c4e9 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a4c53915b22464db4c02633c1180e4b : SUCCESS in 10m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29d5abc40f3d4d67925d09dda836bfc7 : SUCCESS in 10m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90203a121304ba9aecf7d86c55fb882 : SUCCESS in 13m 27s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9689f68582be4cf4bf14a9379ce210c0 : SUCCESS in 1h 45m 48s\n- heat-functional https://zuul.opendev.org/t/openstack/build/784493489dd14a7c96d9764e43213270 : SUCCESS in 1h 15m 44s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/14418357a3fc4fa78b9a912aff72eddd : SUCCESS in 1h 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4af045a07868265126997cef662871a3a0c125ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 10:03:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ced9118fc3c2e0dcdd5aa9e3d0be0a980f95e32c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 10:05:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf7c4e3aae0443e1aa2668c82f5e60f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07aa901b0ed54875ae003100a3086c39 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e91fa22c3cb9f93af08f7597cff2b6cc0065fda","revisions":{"9e91fa22c3cb9f93af08f7597cff2b6cc0065fda":{"kind":"REWORK","_number":1,"created":"2022-08-17 06:22:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/853335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/35/853335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/853335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/853335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/35/853335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/35/853335/1"}}},"commit":{"parents":[{"commit":"d0e2ded71374cffab41499e75efb81782bf83ffa","subject":"Merge \"Remove TripleO job\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d0e2ded71374cffab41499e75efb81782bf83ffa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-20 01:50:26.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 06:22:20.000000000","tz":0},"subject":"Fix AttributeError in Python 3.8","message":"Fix AttributeError in Python 3.8\n\nThe importlib-metadata package was imported to the core Python since\nPython 3.8, and stevedore uses the standard library implementation\ninstead of the 3rd party package in Python 3.8 and later. However\nthe implementation available in Python 3.8 is quite old and especially\nit does not include the change which added the module attribute[1].\n\nThis change makes sure we avoid referring to that attribute if it does\nnot exist, to avoid AttributeError in Python 3.8.\n\n[1] https://github.com/python/importlib_metadata/commit/dfc96137ac3583477c02aa97291b0896fb61f282\n\nBackport note:\nThis change includes an additional fix for Python \u003c 3.8, which uses\nthe 3rd party importlib-metadata.\n\nStory: 2010095\nTask: 45650\nChange-Id: Ieba5f353b0fe8a9b187bac1f6ff1b3a8f3b896b3\n(cherry picked from commit eec1f92bc02e8beb4ce1db9646f8b3b678fb8329)\n(cherry picked from commit ecca0fcbeb96608e56005ec40fe586f498b62f66)\n(cherry picked from commit b8e27f7c47f0f4def7bbd7dc28e6c5c6aba91f9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9e91fa22c3cb9f93af08f7597cff2b6cc0065fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/9e91fa22c3cb9f93af08f7597cff2b6cc0065fda"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[]}
