)]}'
{"id":"openstack%2Fheat~871047","triplet_id":"openstack%2Fheat~stable%2Fzed~Ie12c2df93f846629a7c8b7c6415cebc31eaba1bc","project":"openstack/heat","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14567":{"account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"last_update":"2023-01-26 14:24:15.000000000","reason":"\u003cGERRIT_ACCOUNT_14567\u003e replied on the change","reason_account":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"}}},"hashtags":[],"change_id":"Ie12c2df93f846629a7c8b7c6415cebc31eaba1bc","subject":"Decode UTF-8 body data in SwiftSignal","status":"MERGED","created":"2023-01-19 12:57:15.000000000","updated":"2023-01-26 17:19:49.000000000","submitted":"2023-01-26 17:18:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"871047","cherry_pick_of_change":868256,"cherry_pick_of_patch_set":4,"meta_rev_id":"d4e3300adffdf263765085c81bd4e62213fa7ea0","_number":871047,"virtual_id_number":871047,"owner":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-26 17:18:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-01-23 08:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-01-26 10:02:01.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-01-26 10:02:01.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}],"CC":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-19 12:57:15.000000000","updated_by":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"CC"},{"updated":"2023-01-19 13:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-19 14:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-23 08:40:17.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-01-26 10:02:01.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"}],"messages":[{"id":"25563c4df9662a331d0ae3ea67b83cd09551b648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-01-19 12:57:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2e784f1329c46b97f98e78da8b509e47b33ef3e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 13:31:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a227e77d4459404c898c6c1e9c919c47\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56e7ae5080964f459a3d6e6acbc8ccdf : SUCCESS in 15m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b70ed385c13d4ff7ab46bbf873bbd3cb : SUCCESS in 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91b0d1504bf7d251264c1e87cd70584d0852ca7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 14:25:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3def718fe7f41ebb74ee67466ffc66b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1d19f6057db4cb9b41aea884c224c56 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86daf5f2488a4138b3d1bd6512368ec3 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b825414f1a2441429b1b029dd26331b7 : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6489cf53ed994cd7887909c6a79eb2d7 : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b06d0204d20347499d998c5808de5f39 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/450dde6198db4776982390b68dd6bcc1 : SUCCESS in 13m 12s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/34479f55bb68499ebe3b5fb8c96b9654 : SUCCESS in 1h 23m 09s\n- heat-functional https://zuul.opendev.org/t/openstack/build/6b61a11d57a44d6fbce72e640db51d94 : SUCCESS in 1h 08m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/4ef2edbbfff443f099157020e6d82e1a : SUCCESS in 1h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"16f90a09c8391eebfccd44e6cd49430824b3f401","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 08:40:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6879b108f3cff599525cde20446808485fb70a47","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2023-01-26 10:02:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"55afdf609e6f9e7f7a583a46492f9ad646f7a344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 10:02:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68808d1139391216024470f173b771860f97525d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 11:32:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d490a2cf0a343c4a9a4ec244cc9e930\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff709a5009394015adb808f175a5bbf7 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/447aef616d464aa8892fd327e5767828 : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5e5991d34aa4b648ca3a53e52aeb6e7 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d5a3575e754bf6ad43f07602393a9d : SUCCESS in 12m 46s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/a2e46b7dd513466da827045215463290 : SUCCESS in 1h 22m 13s\n- heat-functional https://zuul.opendev.org/t/openstack/build/fc98872ca09647f5b15c9b0d1ac376a2 : SUCCESS in 55m 34s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/34315c34b70d44a784d24c136541b77b : FAILURE in 50m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e8cf57585d8da8c2a35281f6812766874d6562c2","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-01-26 12:33:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51347c4fe504d97cb005e6c836b20cd309f915fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 12:56:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc6cdf861f5549e2b85ebe4385bb5761\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02ed28202a79465db41e954c2d68a11b : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb86ab4108344777a06b522db58420c5 : SUCCESS in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"90d7b8537b76aede0854808926bd1225a1749192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 13:38:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/439096623d184b60a8afc0dfefed6d2e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e69b92b3f4e4c608f5f4d0287ce8c8b : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e51b2fe038d48c097c9d91c6155660a : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a576dafb0c9e420393ee216e01b02067 : SUCCESS in 8m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aedc83f8c13488ba413482a6aa96e2a : SUCCESS in 7m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f4d1f3954934efe85a6a8d83b52ff63 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b909447028492b9db987c9279a0fb7 : SUCCESS in 13m 20s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/46ed671157dd4466a2096bf1186f2787 : FAILURE in 58m 01s\n- heat-functional https://zuul.opendev.org/t/openstack/build/59e06912430e4a2eaf8b52b59c662a23 : SUCCESS in 50m 36s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/b56d56c29754457ea0ec717ab49e767b : FAILURE in 57m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9c6906149254bd1e4689dfff821529024bcb946a","author":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"date":"2023-01-26 14:24:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e48ccbdb480e33a3f592b791bd388fa999d07c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 14:32:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e82d160b10e48639b4eaaafb970ee9d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f3cbcb21d474b08b1f971a2b83564d6 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d36ef4fe190f475e9ebffa9ca27a5a06 : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"016f41170570c20707aff11b9a982d9ec1c914b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:56:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5493ef3b35244aadb03a0aec52d51d22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7f5b0d5a51f4a259213b5433faacab2 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa06df1272a4767950291d50e7f2662 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b51c35eb6c9e4dafb339989a61e0c462 : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/114f00540eae488e81578f83e06c9081 : SUCCESS in 9m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1281c6e6e1ad4071baf463a9741d1fc5 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29bc8043fdbf4c409e5da2c4194518a5 : SUCCESS in 11m 46s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/afa30a07208d4ea1abee4d1c2619fb2f : SUCCESS in 1h 23m 02s\n- heat-functional https://zuul.opendev.org/t/openstack/build/012b71331e5f413cb1248613dafffb44 : SUCCESS in 1h 09m 24s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/14e65a4a077a4cab8974bde0a9317ff0 : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":1},{"id":"2574b6b8e9fa28dc360c8fa40c09a01d78d3e6e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 15:57:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2e6490b4087b74192a2f08ca4ff48c4f46eb7b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:18:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74976073097c4fdb87de257cf2c96093\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66bd1c8e59da406699a7c74e7b13ac00 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a4b29b244534ffb9c99293baeb0e5fb : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f02f40b494bd4be298901a8f9876c998 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2ae8e2820a49eaa45491e3bd62ed60 : SUCCESS in 12m 02s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/252a53d2ac874cefab11731bf4ce38c4 : SUCCESS in 1h 19m 43s\n- heat-functional https://zuul.opendev.org/t/openstack/build/76c29b68d6df4f968e50e73608489088 : SUCCESS in 54m 45s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/4a7dc0425cbd437cb7f126d95831d9bb : SUCCESS in 50m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8e31f5db70f3c924121649f52681589d2f2eb804","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:18:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4e3300adffdf263765085c81bd4e62213fa7ea0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-26 17:19:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d258b9bf454570a25de30e0e9bc4f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb34de7023a40c2a5d6c3a7a65a1036 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6343b64e703221402f0d40fc0e62da944551e2dd","revisions":{"6343b64e703221402f0d40fc0e62da944551e2dd":{"kind":"REWORK","_number":1,"created":"2023-01-19 12:57:15.000000000","uploader":{"_account_id":14567,"name":"Rodrigo Barbieri","email":"rodrigo.barbieri2010@gmail.com","username":"ganso"},"ref":"refs/changes/47/871047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/47/871047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/47/871047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/47/871047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/47/871047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/47/871047/1"}}},"commit":{"parents":[{"commit":"af617facc5d6a63f416c1a047c91a66008ee4364","subject":"Ignore resource_data decryption errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/af617facc5d6a63f416c1a047c91a66008ee4364"}]}],"author":{"name":"Zane Bitter","email":"zbitter@redhat.com","date":"2022-12-21 03:51:55.000000000","tz":-300},"committer":{"name":"Rodrigo Barbieri","email":"rodrigo.barbieri@canonical.com","date":"2023-01-19 12:57:15.000000000","tz":0},"subject":"Decode UTF-8 body data in SwiftSignal","message":"Decode UTF-8 body data in SwiftSignal\n\nChange-Id: Ie12c2df93f846629a7c8b7c6415cebc31eaba1bc\nStory: #2010485\nTask: 47053\n(cherry picked from commit a0ff4b98b700c7f31a9f14d3d62eeaf3b2047139)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6343b64e703221402f0d40fc0e62da944551e2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6343b64e703221402f0d40fc0e62da944551e2dd"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}}]}],"submit_requirements":[]}
