)]}'
{"id":"openstack%2Fheat~872059","triplet_id":"openstack%2Fheat~stable%2Fzed~I024cd134577338fc1075e7742df7f006dc914646","project":"openstack/heat","branch":"stable/zed","topic":"bug/2003993","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-03-02 07:46:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I024cd134577338fc1075e7742df7f006dc914646","subject":"Fix setting the tempest virtual env constraints env var","status":"MERGED","created":"2023-01-28 04:02:33.000000000","updated":"2023-03-02 07:47:08.000000000","submitted":"2023-03-02 07:46:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872059-bug/2003993","meta_rev_id":"845ff8804c822377b5788ef281807ac099550034","_number":872059,"virtual_id_number":872059,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 07:46:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-02-02 02:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-03-02 06:30:01.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-03-02 06:05:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-03-02 06:30:01.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-28 04:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-01-28 04:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-02 02:33:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-03-02 06:30:01.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"}],"messages":[{"id":"567b81c568735298c749e52c6ed01c9b32676253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-28 04:02:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d15c30947cf449c7d162f854e46edbd6c74dd3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 04:10:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac9617b97024f78b3175073b60d1d23\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d841d5161e6406dad5fdb876aa2e8f4 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b16c4f18509e47c79f78e3bddf6a2d50 : SUCCESS in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b40b18f1a6509f13a6529421c446c6765a86fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 04:48:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43ce3c6836504424b6203853810d7c27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f3bc2076a3a4a38b5f55caab30b79f0 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f31fffd1b014593a5bfad54989ab4f5 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a4828d803c345a5a873d1d2badb8fe7 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a65d1ddcdf3c422b9106cd6a89cbb602 : SUCCESS in 7m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f952abb069fd4ded9e4a4673dfc15f87 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29be028d8ea2413eaed0408f379f1de4 : SUCCESS in 11m 42s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1ed108d9be484d1aa6c3b23d8da980d7 : FAILURE in 43m 43s\n- heat-functional https://zuul.opendev.org/t/openstack/build/15f608a4de6443f9bd39d68f84c82892 : FAILURE in 29m 02s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a20822c2acc9497db424d70b8c21e98d : FAILURE in 25m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6db2fc74b969b62607f2282a646cf2136cae0df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-28 22:27:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0f430c645ad3636c983080f31a10779373bb85f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-28 22:29:46.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3a7e23eeb5652353740b90bb1964bfd918fb562a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-28 22:30:38.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"70346255af5bc9579f86a8ad124f72dc556793b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 22:48:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6eda2aa82ba4682bc4bc0eb7f7e6ad4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a74c1cbd1f84b9cb52f2464837d8a6b : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b1e58d3a4f94f7abf43ac64a32fbc29 : SUCCESS in 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0a23ffc08851780148097b3524b5c82728d4c9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-28 23:50:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b0ce4e0465d4d0abb935824a13fa41b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/191573fd5ee94400912fd834c20397f4 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1976b9d0a54247dd93baeabf8d29bbe1 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0b6de683f3a4badb4971fe8efbd3394 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e92b2aabe484416a97c9b6cca8fecc3 : SUCCESS in 9m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36de6c22fb5f4e8384ce4aab004e8467 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/680fcc47a8544202a966f57174710e13 : SUCCESS in 11m 26s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9b2dbc125b1a4dc5bf912586aee766ea : FAILURE in 43m 46s\n- heat-functional https://zuul.opendev.org/t/openstack/build/4747c525a0244be68c4f7c94f6fe7f71 : SUCCESS in 44m 43s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/b66eff7433f0432ca1850aa8f846fbc5 : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":4},{"id":"9d8418307ddb3d4370cc7692dbb8deee868498d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-29 03:35:15.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"feba3a3d3412253553c7a2a86b28640247e92384","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 03:42:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6eba74001b74dd48a53cb3b3f7b4312\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/457f849aa4b24aa7b88c2bb7687c72f2 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74f4d7637b2a4b3d86a018f70baeeb46 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a39eefacb581488c664972248c15ed913e27d795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 04:49:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f82192e055b4397871c10a114569bd6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea3270ec776477895f335fb1589b42c : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d417f3c73b3d4895add72b7ed4edc5d9 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3691d415210c493f9c8efaeff778e9c1 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3abb70127d824ddcb18c4b79b0741093 : SUCCESS in 8m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8df38bb006b14d2789b1f3dc745b085f : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beec35690e934eefa8acd5be454b50e0 : SUCCESS in 9m 52s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0b2adbf4626c471491eb20c334184a61 : SUCCESS in 1h 08m 46s\n- heat-functional https://zuul.opendev.org/t/openstack/build/f13c9240ff944d53a1b4b2c4e4c1a96f : FAILURE in 1h 09m 59s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/210a88eba6af4b9c9a6d81001cfdb43b : FAILURE in 57m 49s","accounts_in_message":[],"_revision_number":5},{"id":"d55196d3c0fb21bfdcf4fe11a8a86a1271871521","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-29 05:02:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ac58d2ef2f6cbc8249733fe2e404aaea4606d10b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 05:10:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e6547fdbc04230a8bc483895c3139c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fea3d5f96b3347a7bfb1a32ded14bebd : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eba52e883c164aad9f8d014f57a7752f : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3b8f0ec533ffe2706bf1e2b54d38b934cd087173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 06:19:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37c12e454aee4a499ff1194cc26c767c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e9636cc42fb4419a2145536cfea1ec0 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194c51c20f9640b3b3ab51bbd68a89b5 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/508b12fefa254f1f930c188c2b965ca2 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b7c7abbee3f43d9947139bd5d33b50e : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/718def1badaf49058d76a172d4ea8569 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd5b9136e177468a9656179c5547c1a2 : SUCCESS in 9m 50s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/228344d0254c4b318a3e1c2b9f03e94e : SUCCESS in 1h 08m 03s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7e7997854c69419f84b7a356f7f10967 : FAILURE in 43m 22s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/a79735ab1dc04eb89133e82ebfa69d23 : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":5},{"id":"36111b35c6b7c12515308215519dcdd6fe612539","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-01-30 02:38:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"35802bcb3f371123a5b72d98a289c907d05260b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 02:48:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2670773b87ec4b9f8de5a4f8fad7872d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93f4846c75bd47db9800262ad7d83330 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8201886e87a84e9a998c02fc7a704bb6 : SUCCESS in 8m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a712fd48ecf8e504119bc1711c77baf3f080c933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 03:58:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b739e5011dd4b90b2b229ca0bbe374d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d35b6c827a044942a9edd1dab54525b0 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3517a8d042545568e9ef51c9632f7ad : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adba4733ea0d435f887e2e984891f080 : SUCCESS in 9m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13915f090810433ca878c69d05a70498 : SUCCESS in 9m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b9063ad848a4bf2b37b072450655899 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462db15803d749b8bdc44e74ce812219 : SUCCESS in 10m 21s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/14e3cc8f381140fcba1dcca8a7e0ae74 : SUCCESS in 1h 16m 20s\n- heat-functional https://zuul.opendev.org/t/openstack/build/cfe1b7313c1042b98ce068ef706c0db1 : SUCCESS in 55m 20s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/75294e34b1c149f9a7af5eba9420489f : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":5},{"id":"3c758e6d77f698dfc01a1b078c52b2cc2a11bd67","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-02 02:33:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f50a1f1c5837ec8ae42397a9a3e625d8f3d9ed14","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-02 06:05:19.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"be864e6ee6818bc7305fa3e586f0c9033c840a33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 06:05:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"21c83ed09dcb2850757c54a4bc13f022838b49de","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2023-03-02 06:30:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e5a82166cd517d790e41bc5c2e00b49918e554f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:46:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dbe4206c92e414ca1c71ce67baa032b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edfbd5b4f764e5ebbbba6f4cbc6627d : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9040e620c474b0a9e174aa1a07703d6 : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85ea0b8ff1444223923eac47de649ac4 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d1d03c5ebd47a8a51752eca14a72b6 : SUCCESS in 12m 26s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/87e5ae8dfa3b42f184537f2af5ee1d9f : SUCCESS in 1h 38m 58s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b61edc45252c4f74babe03f507bc8065 : SUCCESS in 54m 07s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/dcb23e9ab36b425ba4939dcbc6a2e7b8 : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":5},{"id":"ef9c1cc9d30973ba81fe046ef9dc48b816271181","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:46:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"845ff8804c822377b5788ef281807ac099550034","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 07:47:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/066ebdf30a3a43aa88b624ce3704ae52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f4d70beacc45af8180f54673efc513 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b29266398fbb4e623faaa6a894de2a33bb2846a7","revisions":{"0112e5b58e63008c8ead7fe66d12116367275fa1":{"kind":"REWORK","_number":1,"created":"2023-01-28 04:02:33.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/872059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/59/872059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/59/872059/1"}}},"commit":{"parents":[{"commit":"6343b64e703221402f0d40fc0e62da944551e2dd","subject":"Decode UTF-8 body data in SwiftSignal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6343b64e703221402f0d40fc0e62da944551e2dd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:59:38.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-28 04:02:25.000000000","tz":-360},"subject":"Fix setting the tempest virtual env constraints env var","message":"Fix setting the tempest virtual env constraints env var\n\nHeat upgrade script set the env var TOX_CONSTRAINTS_FILE/UPPER_CONSTRAINTS_FILE\nwhich are used to use the constraints during Tempest virtual env\ninstallation. Those env var are set to non-master constraint when\nwe need to use non-master constraints but when we need to use the\nmaster constraints we do not set/reset them point to master constraints.\n\nThis create the issue when this grenade script install the tempest with\nstable constraints but with master Tempest and as there is mismatch of\nconstraints now with fasteners of stable branches. Below is failure\nof heat greande job on stable/yoga\n\n- https://zuul.opendev.org/t/openstack/build/3aaec4d59bb84068bb4d4428ea747cbd/log/controller/logs/grenade.sh_log.txt#3245\n\nSimilar way, role should set stable constraints only for the stable EM branch\nnot for all otherwise it fail when constraints/requirements are bumped.\n\n- https://zuul.opendev.org/t/openstack/build/74f86b8097f44c35acaffdcfe41d9693\n\nWe should set/reset those constraint env var to master constraints if configuration\ntell to use the master constraints.\n\nCloses-Bug: #2003993\nChange-Id: I024cd134577338fc1075e7742df7f006dc914646\n(cherry picked from commit 824e1e182d9122d775e9d5432368ccd21a0769f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0112e5b58e63008c8ead7fe66d12116367275fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/0112e5b58e63008c8ead7fe66d12116367275fa1"}]},"branch":"refs/heads/stable/zed"},"06536783c9025a0bf516fc608943d7b2c9b5dee6":{"kind":"REWORK","_number":2,"created":"2023-01-28 22:27:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/872059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/59/872059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/59/872059/2"}}},"commit":{"parents":[{"commit":"6343b64e703221402f0d40fc0e62da944551e2dd","subject":"Decode UTF-8 body data in SwiftSignal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6343b64e703221402f0d40fc0e62da944551e2dd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:59:38.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-28 22:27:40.000000000","tz":-360},"subject":"Fix setting the tempest virtual env constraints env var","message":"Fix setting the tempest virtual env constraints env var\n\nHeat upgrade script set the env var TOX_CONSTRAINTS_FILE/UPPER_CONSTRAINTS_FILE\nwhich are used to use the constraints during Tempest virtual env\ninstallation. Those env var are set to non-master constraint when\nwe need to use non-master constraints but when we need to use the\nmaster constraints we do not set/reset them point to master constraints.\n\nThis create the issue when this grenade script install the tempest with\nstable constraints but with master Tempest and as there is mismatch of\nconstraints now with fasteners of stable branches. Below is failure\nof heat greande job on stable/yoga\n\n- https://zuul.opendev.org/t/openstack/build/3aaec4d59bb84068bb4d4428ea747cbd/log/controller/logs/grenade.sh_log.txt#3245\n\nSimilar way, role should set stable constraints only for the stable EM branch\nnot for all otherwise it fail when constraints/requirements are bumped.\n\n- https://zuul.opendev.org/t/openstack/build/74f86b8097f44c35acaffdcfe41d9693\n\nWe should set/reset those constraint env var to master constraints if configuration\ntell to use the master constraints.\n\nCloses-Bug: #2003993\nChange-Id: I024cd134577338fc1075e7742df7f006dc914646\n(cherry picked from commit dbbba907d396158ce3bf3a207a8875b9fbb5a5b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/06536783c9025a0bf516fc608943d7b2c9b5dee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/06536783c9025a0bf516fc608943d7b2c9b5dee6"}]},"branch":"refs/heads/stable/zed"},"2a45850f3e2ace84e48c53e9f869831972f2ef45":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-01-28 22:29:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/872059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/59/872059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/59/872059/3"}}},"commit":{"parents":[{"commit":"6343b64e703221402f0d40fc0e62da944551e2dd","subject":"Decode UTF-8 body data in SwiftSignal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6343b64e703221402f0d40fc0e62da944551e2dd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:59:38.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-28 22:29:35.000000000","tz":-360},"subject":"Fix setting the tempest virtual env constraints env var","message":"Fix setting the tempest virtual env constraints env var\n\nHeat upgrade script set the env var TOX_CONSTRAINTS_FILE/UPPER_CONSTRAINTS_FILE\nwhich are used to use the constraints during Tempest virtual env\ninstallation. Those env var are set to non-master constraint when\nwe need to use non-master constraints but when we need to use the\nmaster constraints we do not set/reset them point to master constraints.\n\nThis create the issue when this grenade script install the tempest with\nstable constraints but with master Tempest and as there is mismatch of\nconstraints now with fasteners of stable branches. Below is failure\nof heat greande job on stable/yoga\n\n- https://zuul.opendev.org/t/openstack/build/3aaec4d59bb84068bb4d4428ea747cbd/log/controller/logs/grenade.sh_log.txt#3245\n\nSimilar way, role should set stable constraints only for the stable EM branch\nnot for all otherwise it fail when constraints/requirements are bumped.\n\n- https://zuul.opendev.org/t/openstack/build/74f86b8097f44c35acaffdcfe41d9693\n\nWe should set/reset those constraint env var to master constraints if configuration\ntell to use the master constraints.\n\nCloses-Bug: #2003993\nChange-Id: I024cd134577338fc1075e7742df7f006dc914646\n(cherry picked from commit dbbba907d396158ce3bf3a207a8875b9fbb5a5b1)\n(cherry picked from commit 06536783c9025a0bf516fc608943d7b2c9b5dee6)\n(cherry picked from commit 43e493fb9ea773b3f9c767ddf53711ebcdfb6a58)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2a45850f3e2ace84e48c53e9f869831972f2ef45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2a45850f3e2ace84e48c53e9f869831972f2ef45"}]},"branch":"refs/heads/stable/zed"},"388f0aadf349bc3183863c8d34e98c8f84a6c9b7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-01-28 22:30:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/872059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/59/872059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/59/872059/4"}}},"commit":{"parents":[{"commit":"6343b64e703221402f0d40fc0e62da944551e2dd","subject":"Decode UTF-8 body data in SwiftSignal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6343b64e703221402f0d40fc0e62da944551e2dd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:59:38.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-28 22:30:27.000000000","tz":-360},"subject":"Fix setting the tempest virtual env constraints env var","message":"Fix setting the tempest virtual env constraints env var\n\nHeat upgrade script set the env var TOX_CONSTRAINTS_FILE/UPPER_CONSTRAINTS_FILE\nwhich are used to use the constraints during Tempest virtual env\ninstallation. Those env var are set to non-master constraint when\nwe need to use non-master constraints but when we need to use the\nmaster constraints we do not set/reset them point to master constraints.\n\nThis create the issue when this grenade script install the tempest with\nstable constraints but with master Tempest and as there is mismatch of\nconstraints now with fasteners of stable branches. Below is failure\nof heat greande job on stable/yoga\n\n- https://zuul.opendev.org/t/openstack/build/3aaec4d59bb84068bb4d4428ea747cbd/log/controller/logs/grenade.sh_log.txt#3245\n\nSimilar way, role should set stable constraints only for the stable EM branch\nnot for all otherwise it fail when constraints/requirements are bumped.\n\n- https://zuul.opendev.org/t/openstack/build/74f86b8097f44c35acaffdcfe41d9693\n\nWe should set/reset those constraint env var to master constraints if configuration\ntell to use the master constraints.\n\nCloses-Bug: #2003993\nChange-Id: I024cd134577338fc1075e7742df7f006dc914646\n(cherry picked from commit dbbba907d396158ce3bf3a207a8875b9fbb5a5b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/388f0aadf349bc3183863c8d34e98c8f84a6c9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/388f0aadf349bc3183863c8d34e98c8f84a6c9b7"}]},"branch":"refs/heads/stable/zed"},"b29266398fbb4e623faaa6a894de2a33bb2846a7":{"kind":"REWORK","_number":5,"created":"2023-01-29 03:35:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/872059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/59/872059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/59/872059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/59/872059/5"}}},"commit":{"parents":[{"commit":"6343b64e703221402f0d40fc0e62da944551e2dd","subject":"Decode UTF-8 body data in SwiftSignal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6343b64e703221402f0d40fc0e62da944551e2dd"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-01-27 20:59:38.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-01-29 03:35:15.000000000","tz":0},"subject":"Fix setting the tempest virtual env constraints env var","message":"Fix setting the tempest virtual env constraints env var\n\nHeat upgrade script set the env var TOX_CONSTRAINTS_FILE/UPPER_CONSTRAINTS_FILE\nwhich are used to use the constraints during Tempest virtual env\ninstallation. Those env var are set to non-master constraint when\nwe need to use non-master constraints but when we need to use the\nmaster constraints we do not set/reset them point to master constraints.\n\nThis create the issue when this grenade script install the tempest with\nstable constraints but with master Tempest and as there is mismatch of\nconstraints now with fasteners of stable branches. Below is failure\nof heat greande job on stable/yoga\n\n- https://zuul.opendev.org/t/openstack/build/3aaec4d59bb84068bb4d4428ea747cbd/log/controller/logs/grenade.sh_log.txt#3245\n\nSimilar way, role should set stable constraints only for the stable EM branch\nnot for all otherwise it fail when constraints/requirements are bumped.\n\n- https://zuul.opendev.org/t/openstack/build/74f86b8097f44c35acaffdcfe41d9693\n\nWe should set/reset those constraint env var to master constraints if configuration\ntell to use the master constraints.\n\nCloses-Bug: #2003993\nChange-Id: I024cd134577338fc1075e7742df7f006dc914646\n(cherry picked from commit dbbba907d396158ce3bf3a207a8875b9fbb5a5b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b29266398fbb4e623faaa6a894de2a33bb2846a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b29266398fbb4e623faaa6a894de2a33bb2846a7"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}}]}],"submit_requirements":[]}
