)]}'
{"id":"openstack%2Fheat~873463","triplet_id":"openstack%2Fheat~stable%2Fxena~Ifc51ff6a4deab05002ccded59383416f9a586aa0","project":"openstack/heat","branch":"stable/xena","topic":"bug/2010484","attention_set":{},"removed_from_attention_set":{"35639":{"account":{"_account_id":35639,"name":"ChengEn, Du","display_name":"Chengen Du","email":"chengen.du@canonical.com","username":"chengendu"},"last_update":"2023-04-11 06:44:24.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-11 06:44:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifc51ff6a4deab05002ccded59383416f9a586aa0","subject":"Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command","status":"MERGED","created":"2023-02-13 13:05:48.000000000","updated":"2023-04-11 06:45:40.000000000","submitted":"2023-04-11 06:44:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873463-bug/2010484","cherry_pick_of_change":868166,"cherry_pick_of_patch_set":4,"meta_rev_id":"61917bc75df01968f8abf984690e135c781830b5","_number":873463,"virtual_id_number":873463,"owner":{"_account_id":35639,"name":"ChengEn, Du","display_name":"Chengen Du","email":"chengen.du@canonical.com","username":"chengendu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 06:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-04-11 05:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2023-04-10 08:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2023-04-11 05:27:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 13:22:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-13 13:47:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-02-13 14:23:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 05:27:46.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"efb0a52799a447933e234110e0f6e8fcb0616716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35639,"name":"ChengEn, Du","display_name":"Chengen Du","email":"chengen.du@canonical.com","username":"chengendu"},"date":"2023-02-13 13:05:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4266a7e137ceaffb0c527154458dc33ecf420a7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:22:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae1c1cbcad3a4b39832163c21d880e4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a415fe460234f2ba1481d3efdd7e45e : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36e30f94eaf6405da15dc4273c1543ff : SUCCESS in 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0bdd2f57f0f2f1e6869eaf1fd2d349ee04b0025","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-13 13:47:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"85e1230142606ad566633f7590a0366beedc5b6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 14:23:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2adaa3e7088847f48588b8957c547946\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb01cceb08fc4b54876114bfdcaca719 : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927cda57df8f4d0bad509479ff3176e1 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f8cc1f351e432b948c529ed1b47d55 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/466d6586a8914b318a70fdcb32431da6 : SUCCESS in 8m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07235a6552a14ca482d88b54f30cda90 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07cc166ba3f4456297ca082ef32af98f : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78e807b341604136b0e87b0a540c28e6 : SUCCESS in 7m 09s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0974d63c5ee84618a96d2879c14be27a : FAILURE in 1h 13m 25s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b9830ff884604ad288ac344e829cd7a3 : FAILURE in 24m 30s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/3967b85ffb4649b29719f782e6e3dc85 : FAILURE in 41m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9867b0021c8f65b164303b439cd7d930ab9a6e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-29 11:07:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b35615f42982bbb2f7d272b485de74f0f0ddfe8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 11:38:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76662c4b754c4b7fa9691c028a32d6d1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcb6a20c49f14821adbb4077b5835d70 : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a649edca0f345b0a58e38505088e61c : SUCCESS in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e6a88686d51e79febb1446164ffc69f5bb88384a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-29 12:35:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08273250171b4b3ba77d195e003be941\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7319238772834af38d063de71e40785c : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c9e964f0ff14d5cb8b3fc64c1f16956 : FAILURE in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdc1c4c3cc324a998626111962a76707 : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/891b880aaaaf4f7b8eeebf482c46c26a : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e89077b8587b4af3a6ebd45b05fc1a3e : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0867ed5371c4d07ab88708fdbc6ea70 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb1afac326fa4fc5a74cef0db44d649c : SUCCESS in 7m 36s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/28682a34b3794e298f97ba1cba2f296e : FAILURE in 1h 21m 56s\n- heat-functional https://zuul.opendev.org/t/openstack/build/416da7edadf5425ca9de75c1d4f8078e : FAILURE in 31m 20s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/c07732d144a34cd49e78796e9561472a : FAILURE in 42m 24s","accounts_in_message":[],"_revision_number":2},{"id":"37abd42c241f251f5ec3d3e685b327789884ec07","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 07:12:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"faca6f7799175c8681e97f0cffef86d094385480","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 07:31:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d35eeb9af46435ebf0faf4c1994d121\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab1e18ccb32448409a827f9b5ecbfe0b : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f5ddf5cbee8442792b516e3d7ee21fd : SUCCESS in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be9df7446a504d55d6a1290b850d46a454188f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 08:48:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11c068d18e6947c2b245ac8208bc9db3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30e3cfa02af24bdd8199bc994f2463b1 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65d7d1b3955940b5bbe9b2d9d6fa9110 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/125ffa7a319a440aa3fb33216aa2179f : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ef73b77d64d4a4795e356a437b24aac : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db860ce0440947fdbcb5442fe43ba864 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf108fb044e4f7a82c2430823c8d8d3 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08afe062d91e4ea99b96ced7e4681265 : SUCCESS in 10m 10s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/a49290e18c014efb8bbb97e3c74acec8 : FAILURE in 1h 20m 33s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/c96770821ad04a3caf213c837f4be579 : SUCCESS in 1h 03m 48s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/c42373b179a04526a77e3e9cfe7aa8ca : SUCCESS in 56m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7dfa68843ce3c4bd6314f05f01606f7d95b2ab2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 08:54:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62eda0c1bfac26495ca90e672dd6c1938c3fe353","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-04-11 05:27:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"94c26fc93e540580caaa6781ffe7fde9079705a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 05:28:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f2c7e6cd315ff33d9ff8cdff2c2ce218f0f09f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 06:44:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1323074da76a4303bc5cfa50350aa2a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd098c796ff42d8834f5ea5c3ceadb9 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a55805a59144a34af0d049850a6f826 : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f66fb02cd3204135a1c39700395a94ee : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1793187e1d440ba53892d625595416 : SUCCESS in 14m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b88ccd0eca7947b2b3c1351570bbd960 : SUCCESS in 9m 25s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0f9be75328c64c93abebe4795c72ccf6 : FAILURE in 1h 14m 01s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/d079b4137d6c4c548162e718582643d5 : SUCCESS in 49m 56s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/56b6d9c3f6614b3ba380cae6c98b5024 : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6da1e200ac42146c3fbc65f4666fbb23cd360669","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 06:44:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"61917bc75df01968f8abf984690e135c781830b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 06:45:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b622a79035b644d1aff0b0746341fc6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe0ebc0911446ac88eb6e26d5377c3c : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6edad7cc41340498be764b642863c87 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a49525c547a607743160a33b6a23041378ff177","revisions":{"6821d498fea02584e23a70e0fa5b0ce4928ceb20":{"kind":"REWORK","_number":1,"created":"2023-02-13 13:05:48.000000000","uploader":{"_account_id":35639,"name":"ChengEn, Du","display_name":"Chengen Du","email":"chengen.du@canonical.com","username":"chengendu"},"ref":"refs/changes/63/873463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/63/873463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/63/873463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/63/873463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/63/873463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/63/873463/1"}}},"commit":{"parents":[{"commit":"b135a99575ed42882d17b4bef2006a10cfe8afaa","subject":"Ignore resource_data decryption errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b135a99575ed42882d17b4bef2006a10cfe8afaa"}]}],"author":{"name":"Chengen Du","email":"chengen.du@canonical.com","date":"2022-12-20 10:00:00.000000000","tz":480},"committer":{"name":"ChengEn, Du","email":"chengen.du@canonical.com","date":"2023-02-13 13:05:48.000000000","tz":0},"subject":"Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command","message":"Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command\n\nRelated-Bug: #1999665\nStory: 2010484\nTask: 47052\nChange-Id: Ifc51ff6a4deab05002ccded59383416f9a586aa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6821d498fea02584e23a70e0fa5b0ce4928ceb20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6821d498fea02584e23a70e0fa5b0ce4928ceb20"}]},"branch":"refs/heads/stable/xena"},"1a49525c547a607743160a33b6a23041378ff177":{"kind":"REWORK","_number":2,"created":"2023-03-29 11:07:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/873463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/63/873463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/63/873463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/63/873463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/63/873463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/63/873463/2"}}},"commit":{"parents":[{"commit":"b135a99575ed42882d17b4bef2006a10cfe8afaa","subject":"Ignore resource_data decryption errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b135a99575ed42882d17b4bef2006a10cfe8afaa"}]}],"author":{"name":"Chengen Du","email":"chengen.du@canonical.com","date":"2022-12-20 10:00:00.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-29 11:06:56.000000000","tz":540},"subject":"Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command","message":"Honor \u0027hidden\u0027 parameter in \u0027stack environment show\u0027 command\n\nBackport note:\nThis includes change I0abbd535aacc03446ada0fa806dfdfdaa4522afe which\nfixed the wrong explanation in the release note file.\n\nRelated-Bug: #1999665\nStory: 2010484\nTask: 47052\nChange-Id: Ifc51ff6a4deab05002ccded59383416f9a586aa0\n(cherry picked from commit c1b5fcfbe37d3f94d5b73ae36e071d1aa8c4e0f5)\n(cherry picked from commit 436ecad6511b30683f6b9af4e484ed64c3bcaab9)\n(cherry picked from commit c50d720aa36fd7553b133a5a2f4fefdb3040211f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1a49525c547a607743160a33b6a23041378ff177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1a49525c547a607743160a33b6a23041378ff177"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
