)]}'
{"id":"openstack%2Fheat~876922","triplet_id":"openstack%2Fheat~stable%2F2023.1~I039d1e672611b7b523ce61df072c60d161851eed","project":"openstack/heat","branch":"stable/2023.1","topic":"story/2010638","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-10 12:34:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I039d1e672611b7b523ce61df072c60d161851eed","subject":"Fix unit tests with python-novaclient 18.3.0","status":"MERGED","created":"2023-03-09 13:15:53.000000000","updated":"2023-03-10 13:35:45.000000000","submitted":"2023-03-10 13:34:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876922-story/2010638","cherry_pick_of_change":876950,"cherry_pick_of_patch_set":1,"meta_rev_id":"04ac51bf909c87dd9b72d3f4a08bd99f0be218fe","_number":876922,"virtual_id_number":876922,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-10 13:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-03-10 04:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2023-03-10 04:38:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 13:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-09 13:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-10 04:38:04.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"4055b1048b0638692ece8597a228818bfb4a2561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-09 13:15:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4ca70959e417d4f6faf9939637c71b6ec50570cd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-09 13:16:39.000000000","message":"Topic story/2010638-stable/2023.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"8dd9b359f0f1c533156cf0f6fdedd6fb842066c7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-09 13:16:43.000000000","message":"Topic set to story/2010638","accounts_in_message":[],"_revision_number":1},{"id":"b6373c0da9cae343954fb094ce13aaa889ec990a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:38:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3111318bef1349059e8d71e9137020b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83a4e0bd78844647b4b96e833c18443f : SUCCESS in 18m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49585f67bf514a67af75bdae635b2d5c : SUCCESS in 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f501432e13f66e7d528558c20911758194c1266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:54:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea3575815054c75ab9916d0c4065035\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4bdfcae42d149d7ae21119184089623 : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82401e7f67d447b588cade2fa51f5410 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25106d4c2e9b497ebc01eb0b32c4f6d1 : SUCCESS in 8m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ccf2d158bb34120bcabce5c55abccac : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d3bd4ef375472fb146f45c6d770665 : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a92782d673f4cfe7447d2596f59930659807f566","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-03-10 04:38:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d1dc559f43463c2d5aba6df13214749a7c192de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 04:38:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65dacdd7d4c16cb2ad7207df4f2f517410b49fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 05:12:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfb9e728cee640fea387b51dd2b34206\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c3eb65b1ef4e36988e47e64457aebd : FAILURE in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a72c4b0893148d0a19e4c6b3eb4d644 : SUCCESS in 8m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ddc2454c4854faba6c538812228be9e : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7c7325973441b6b2da89e6a53ffb4e : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2f90fe9951eacd8d96458464615c9e5ee6c755f6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-10 12:34:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d3e1a39eb6f417dc667fb136d22c12e6663c731","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fcedbcee8584095aa7b2f000e128910\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53273760a29b499caae8a6fb592e5d45 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7398674bfe454b78a38f6c12ec982f93 : SUCCESS in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f803686cfb4bb5bc2bcc4eba3db00a8432d6437c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 13:02:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa923c3407c04cadb3aced570a5d6636\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4191b761143b47eead93f27a47bbb8b8 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52e3ea1741494e98b70198e0efb5ee86 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccf53a7f5ef54689a9ce85d31a9e3f72 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/387b40c0397946ac975e4be4fd65daec : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e313a886214e1692910b78fa16185b : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"99803f7915fa2b7bb346d73601ec95414093c878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 13:02:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9acc48b1500c92e60a2ea0d2f97b8191d8c7929","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 13:34:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fde69b905d343b89f622d67be627135\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e33f93ad0b144148e4b9f5dedd1c174 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deb49d0963a9431a8ebc32448b975a34 : SUCCESS in 7m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b69b102b30e2479abe27ae8825853301 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2966bb992a664be8b788a528f150cb4e : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"76e1fad663e852f53925f75d545f9678e43ea824","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 13:34:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"04ac51bf909c87dd9b72d3f4a08bd99f0be218fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 13:35:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e13c5bb88c744d15b6dee6daa5994f2a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a36fd986ad4393ab7d1bda2c1a1424 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cd439a27ac5351b04bea8b6afe7224c704a6bc7","revisions":{"8cd439a27ac5351b04bea8b6afe7224c704a6bc7":{"kind":"REWORK","_number":1,"created":"2023-03-09 13:15:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/876922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/22/876922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/22/876922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/22/876922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/22/876922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/22/876922/1"}}},"commit":{"parents":[{"commit":"e582ef5b0b8cdf96e6af2b90b59a2f2b1c0d4ab0","subject":"Merge \"Return HTTPBadRequest from circular dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/e582ef5b0b8cdf96e6af2b90b59a2f2b1c0d4ab0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-09 10:05:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-09 13:15:53.000000000","tz":0},"subject":"Fix unit tests with python-novaclient 18.3.0","message":"Fix unit tests with python-novaclient 18.3.0\n\npython-novaclient 18.3.0 replaced usage of kwargs by explicit arguments\nand that broke some assertions of arguments.\n\nstory: 2010638\ntask: 47613\nChange-Id: I039d1e672611b7b523ce61df072c60d161851eed\n(cherry picked from commit 513dfd116e5177cb43a0fb78471ad1992f16210d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/8cd439a27ac5351b04bea8b6afe7224c704a6bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/8cd439a27ac5351b04bea8b6afe7224c704a6bc7"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
