)]}'
{"id":"openstack%2Fheat~877646","triplet_id":"openstack%2Fheat~stable%2F2023.1~I3b18de721949f3c15c776f90720f3e994f7d6670","project":"openstack/heat","branch":"stable/2023.1","topic":"fix-wrong-attribute-heat-senlin-client","hashtags":[],"change_id":"I3b18de721949f3c15c776f90720f3e994f7d6670","subject":"fix wrong attribute in heat senlin client","status":"ABANDONED","created":"2023-03-17 06:48:28.000000000","updated":"2024-09-26 16:56:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":876822,"cherry_pick_of_patch_set":8,"meta_rev_id":"fad36608e7418831f0afba1ec60f9e8060ce16a7","_number":877646,"virtual_id_number":877646,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-17 07:56:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 06:48:28.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"state":"CC"},{"updated":"2023-03-17 07:00:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-17 07:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e96d3bac82d891954cb9fb4772943db36a381a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 06:48:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2a2128d340cffd206f1661aae29a138a11d7eade","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 07:00:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a83b43569094decb17203326d33224b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6c757121f7e4a96b51de3898e477acf : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d377fd00b134fe9bdef9d47b0b5c0b2 : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee2647dc4988b829fefd9ecf50b8d9cd9299ef84","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 07:04:24.000000000","message":"Topic fix-wrong-attribute-heat-senlin-client-stable/2023.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"92cd34ccfcbd80c18b890a163fc9530af18cbae2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 07:04:27.000000000","message":"Topic set to fix-wrong-attribute-heat-senlin-client","accounts_in_message":[],"_revision_number":1},{"id":"a53ada3267a377fecd3d68f09b73967ebc29740f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 07:56:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7dcbdd798114d3885165409ca6b4b22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3599fee46ef4053b93b424b0bd8a7d6 : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07085f14e3464f23afcac931cf89ff35 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd88ca9f5c8e41e98cdd12419321813e : SUCCESS in 8m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/793270396e1b423da0d9d09eb7ef9654 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b54177ee2304d0485f1a095116711fc : SUCCESS in 11m 44s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b5664cf866634a6f871aa8facbd7fee3 : SUCCESS in 1h 05m 51s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/4a6c825a5e09468f8739ffd8ff57b3af : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"fad36608e7418831f0afba1ec60f9e8060ce16a7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-26 16:56:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b069b352a26d3ff07406198d3097de70e617ee24","revisions":{"b069b352a26d3ff07406198d3097de70e617ee24":{"kind":"REWORK","_number":1,"created":"2023-03-17 06:48:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/46/877646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/46/877646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/877646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/877646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/46/877646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/46/877646/1"}}},"commit":{"parents":[{"commit":"6a4c7b57b8383bef4a02fb485e781c4cbb2a03d8","subject":"stable-only: Pin bandit to 1.7.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/6a4c7b57b8383bef4a02fb485e781c4cbb2a03d8"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2023-03-08 06:31:12.000000000","tz":420},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-17 06:48:28.000000000","tz":0},"subject":"fix wrong attribute in heat senlin client","message":"fix wrong attribute in heat senlin client\n\nHeat has been using the deprecated interfaces of openstacksdk for some\nSenlin APIs but these interfaces were already removed. This fixes that\nproblem and replace the removed interfaces by the new ones.\n\nstory: 2010635\ntask: 47602\n\nChange-Id: I3b18de721949f3c15c776f90720f3e994f7d6670\n(cherry picked from commit cd2d8dcf08f0fccd2371ad8d50b4034d77bbc038)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b069b352a26d3ff07406198d3097de70e617ee24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b069b352a26d3ff07406198d3097de70e617ee24"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
