)]}'
{"id":"openstack%2Fheat~879014","triplet_id":"openstack%2Fheat~master~I2f6fb33730c083e8cc4d8e0d7ae469be129b5b9c","project":"openstack/heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-10 09:18:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2f6fb33730c083e8cc4d8e0d7ae469be129b5b9c","subject":"DNM: Re-enable a few flaky tests","status":"ABANDONED","created":"2023-03-30 05:37:04.000000000","updated":"2023-04-10 09:18:36.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"72d468157662e44008a3c4145e6be88145b1f85d","_number":879014,"virtual_id_number":879014,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-03-30 07:12:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-30 05:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-30 07:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0be5545440d4d5a093cf583621d5da0083f18b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-30 05:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12292e04f3984c3f952c2ababef57228e218dac7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 05:44:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/564c24d2d3bd47509a1cb6bb7d03d99a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a543af40125947229eb8e2f0072cb56f : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c3dc8e9a1b44f19a35cb8f30702fdd5 : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73ae67fd94943d6cdd6f5c2a7eb588321f857e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 07:12:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32cc1ff405ef43df95af58a9aef2a7ad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d93de672a842cabca0072d6dd9cd4f : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f02c8e4ae3b4f48abf9c72d48d3f2f3 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9c01bf442b445cfa0cf008aaa346032 : SUCCESS in 8m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09cd7cde504d4babb8fc77a99bc928f3 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200d27bed30f452ba70ffc1d5c06ad8f : SUCCESS in 11m 37s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/54f8867abdd6466b84b2f2392d89972e : SUCCESS in 1h 05m 18s\n- heat-functional https://zuul.opendev.org/t/openstack/build/6f7f4486b9db48fb9819680a8843901d : FAILURE in 1h 14m 43s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/39195893fb9b4a8bac87a2d8abdbdc53 : FAILURE in 1h 01m 09s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5bfed63044447a5a1d6522b95f06479 : FAILURE in 1h 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2a9ca529ae4692f998af912e3cb99868b1a3f89","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-30 07:46:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f308bb56729d8103e86c08e29711f9bb2ffbaa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 08:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817ffddf27de4550b8e95834ddd048f0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6dc454368254f44b4cc3d31f9f3cfdf : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30bbec7c5d384314a3b468c4d28da358 : SUCCESS in 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12004db468c71c32a6f246fe5c989560a3ffc497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 09:14:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59534ed626be4a0284eda172745e5079\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2e5f5edb5a34baf9369b23af97459bc : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de5bfc9eb854bc896f50b072e757585 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a392e84e9c54b2db2749c4ee1d2588b : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/873ec844cd9748dc8f20a9f964fad383 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff56d9d07f144e08ac2592f1882071b : SUCCESS in 12m 26s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/59f1235d930e46b19d935d25ed5c30df : SUCCESS in 1h 22m 43s\n- heat-functional https://zuul.opendev.org/t/openstack/build/26adcc52e6524a9eb63612afc70e2744 : FAILURE in 1h 04m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/88de18f2bea647449065272504f868ec : FAILURE in 1h 08m 15s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf34f9d6a1aa43c79ffcebc8b9f787ae : SUCCESS in 47m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d3e876edbdfff5cac76ced4a469f7bccb4cc9b1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-30 10:48:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60edf48ef85a41169b47949184440a2d2e4f468b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 10:58:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e632ea99811424baaabbe81c7449f24\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a154f1ffc5704dd782a1d68d22f716d6 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/503908b8505848de8f83182601793784 : SUCCESS in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b5ae3c4b2b2e7c71102f88a86655b5dfcd1f8432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 12:31:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/507398bf337c46f0bc5e7de0aa3925d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f51dadfa0d0a47149674705d55ff97a3 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218ce296d41444e2a76fd57d33978e3e : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcbfbd1baf404dcb9871d23978eda8ec : SUCCESS in 7m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1417be710c84c13b8069fe37ca07681 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af57abdff4a4595bece1834cc81dbf2 : SUCCESS in 12m 38s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/2ccbdfedaa02462b91eb1e2592677993 : SUCCESS in 1h 30m 13s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c6fa7e202bc546d4a98df20c62819f0d : FAILURE in 1h 10m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/e4810d855c1a4ea8be0a7b27f444900e : FAILURE in 1h 34m 58s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/04367dbdd5cf4166b8b41aa62aa7aa80 : SUCCESS in 1h 26m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0bf84fc232b99a29d22d9d7d2561e7134f7df90","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-30 13:18:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"128657aafc4e6e03d423c61566a0dd458a12a4c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:25:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a61dddd758284fc8b104257929341b47\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fe1ded108ce4c148176cab3883e13b0 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/065792180913439e911c2baeaf638ea0 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d32feec893f30263db84d1bc891d15a38e76d3c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-30 14:30:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"020daed0365ae31bf084176912827d1af4fc672d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 14:41:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/459885cdc4c74462a7337d88bb4fdeb4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/519c0d23b9c8449b8838d3c722b24a4b : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0dc58953f4449308f431b0419dc1179 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f06b5b0d010b45c3a6bdbf25a8dbd9884f16ed35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 14:54:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a955d55810c34e11acabb151e6a66f70\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cb02547e65846c69ded76ded047e54d : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2293f7f8db4f75866229dcaa5a2d6f : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f127ca69539c491fbd7a267ed7b401e2 : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9be9b3b21c0b42a88c46f34974c1f49e : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b85b424b1794438b3ddcc84656468b3 : SUCCESS in 13m 27s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/0a56386e111d478686cfc8bcc66b755a : SUCCESS in 1h 26m 47s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b35fcf183d5c42bfae9a1527707c518c : SUCCESS in 1h 02m 25s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/1e977c10cb1147ba9b646056ac6b55e7 : FAILURE in 1h 12m 49s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/c645349a93af42ef8b83b70678175dd7 : FAILURE in 52m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e643a9894a0b6b646c05231d4d3fb2a8e7029802","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-31 01:14:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7d6ce5a66018df1c537c3d603faa1bcd674a974","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 01:23:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/500656db835d48f689760feda8b73a1a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fcd5c67761943f4ad5226746a352510 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc907972596e444da4a23c0ac062b1d8 : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04a6ed8f5fd0ae2e53199cd3bbe8cdedcda61481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 03:04:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e773f8daa564e218a9652841ebe9529\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/424a319eae304b15979de10f7999fcda : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18d985a411b40f2886df72417dca8e6 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bba80c2f7874e6dbc8c018f16309d16 : SUCCESS in 9m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45c654c1c3ee42c48c2e5a7192480b5b : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804e03182e564fdb9d9eb27cdf756282 : SUCCESS in 9m 23s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/eee0200705b74eeca65f855cc448f66c : SUCCESS in 1h 43m 54s\n- heat-functional https://zuul.opendev.org/t/openstack/build/157bf355ce0242c3a5ef04548a4ee8a0 : SUCCESS in 44m 24s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/aaf5b282a55a4da3b307a920422d5ea7 : FAILURE in 1h 04m 59s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/8eb330694db04fc1a628ff124939a8d3 : SUCCESS in 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4191385737d856fc85ffb30cc77a0508aa198ad0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-31 04:39:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48db082315a5608b54fc81ba88db5a1503add30e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 04:49:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80e8658cdb9b40b5ab5a499035350013\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfe12fb03e9b40aa87346fc7cbe3a4b3 : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8fd73aea075040d78c9dc2c3908c11b8 : SUCCESS in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b1c8f173aaa5d6a9d96bdea69b2ec6bcc933aaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 06:13:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8aa37ce85e8486f97da79cd9fdcdcac\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15fd5a5064914354b4f2a32430a18309 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9006381bcac43c99d5a2249126aac99 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f1f7489e7cc4bdb8b323020d1b767dc : SUCCESS in 8m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7755e5ff4bc455ba2964b0e1a9e5fa7 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03756fd8bdd846f18c98e4f9782806ed : SUCCESS in 12m 50s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/36b7e9987e3f434fae0949c4ff5fb66b : SUCCESS in 1h 27m 38s\n- heat-functional https://zuul.opendev.org/t/openstack/build/897aabf16ef3443a9cd9a0e198a3ab1f : SUCCESS in 54m 23s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7bb0641962194201b5cb4a33943aa956 : FAILURE in 1h 02m 11s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfc410d216de490ba2100c2bbeebd4a4 : SUCCESS in 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72d468157662e44008a3c4145e6be88145b1f85d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 09:18:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"344e08e2661a463c625af1d9b8e0177638df08eb","revisions":{"344e08e2661a463c625af1d9b8e0177638df08eb":{"kind":"REWORK","_number":1,"created":"2023-03-30 05:37:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/879014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/14/879014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/14/879014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/14/879014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/14/879014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/14/879014/1"}}},"commit":{"parents":[{"commit":"5c7862dc7b01c9f610b7513074b23311278d6385","subject":"WIP: Add CentOS 9 Stream job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/5c7862dc7b01c9f610b7513074b23311278d6385"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-30 05:36:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-30 05:36:49.000000000","tz":540},"subject":"DNM: Re-enable a few flaky tests","message":"DNM: Re-enable a few flaky tests\n\nChange-Id: I2f6fb33730c083e8cc4d8e0d7ae469be129b5b9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/344e08e2661a463c625af1d9b8e0177638df08eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/344e08e2661a463c625af1d9b8e0177638df08eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
