)]}'
{"id":"openstack%2Fheat~886549","triplet_id":"openstack%2Fheat~master~Ib8a6a1087c2a33155c7471e641ede643cd250852","project":"openstack/heat","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2023-09-22 16:48:54.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-22 16:48:54.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-22 16:48:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ib8a6a1087c2a33155c7471e641ede643cd250852","subject":"db: Disable autocommit","status":"ABANDONED","created":"2023-06-21 01:45:38.000000000","updated":"2023-09-22 16:48:54.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8f2f88f05234bebf0fb3ab3bd3c79a0eb60244f3","_number":886549,"virtual_id_number":886549,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-22 07:29:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 01:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 04:49:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 13:55:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"0885d8bbf44b957254e24af88d621b5075056a94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-21 01:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b69ceaae4b895b273990916d8b5eaa1eaa93f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-21 01:46:29.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0b13da9ebc2dfae3085a630f278bd83ba906c450","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 01:55:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fffeb7641a5644ab8a6a8cd8624becee\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a8b70dc44184617a4b2df18733bc9d0 : FAILURE in 6m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/846dbf49655e44de8da7eeed73d55a31 : FAILURE in 7m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/727c066a097a473eb4241cfdfe17cbfe : FAILURE in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb3be92998a118ae157e80d06ee14401d5eaccf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-21 02:10:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e12024755eca8fd8daf8251f453b17da70b174c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 02:23:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25e392cd31a444a4b846270abdda720b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af1c09c6322a4621bc5d3824a3a27f3d : NODE_FAILURE Node request 300-0021502527 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53646d039e8b40c0a45bb4628afcfcab : NODE_FAILURE Node request 300-0021502528 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/926041e68c924d51b552dc97b69891bd : FAILURE in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"813ba6b093a4fbd878398298546c46f399233147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 04:49:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b0e0a6c6cbb43d9b43cb5722ed7b040\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccfc9d80b265452d88c586c96343fe11 : FAILURE in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3e463d888c49a79f7c51b3572be219 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a1f9c84f6548628aafbba39cf3caad : FAILURE in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28a624a39ea24180809ee799910ca726 : FAILURE in 7m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/184f47eb37344d4b8098212d13a57b7c : FAILURE in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45bfaf7cc9cf4112a45b5ca3d85c388b : SUCCESS in 11m 48s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/2d49f2fb241a4c39bd17a676c94f1176 : FAILURE in 1h 20m 34s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7b193f10b8e04a12a5a5852758682255 : FAILURE in 59m 02s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/477b039a57bd43a5a2130bd2c41cc9c5 : FAILURE in 1h 21m 25s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ea8b772e8524e68bfb1fdb38e968c75 : FAILURE in 1h 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"330495e2d993f91621c502d4998b6024c126c936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-06-22 06:06:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4f7c292c0ee61d924823588b6f068b98ba9c7fd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 06:15:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95756f215446426c926902f9f8e93d1a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddd33e1cd81a44cda31750943d06b2e9 : FAILURE in 8m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e47e1ae1352147c2a8bb708d34d32eaa : FAILURE in 7m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9080cae5b1b244c6905b1d64191a4140 : FAILURE in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"20c00fdd304de303aa2063f5bdc44dcc49867ade","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 07:29:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c037213b33b645d4bc784d856d1ff581\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f53cbe7d32ec44048c3ecc50afc54bda : FAILURE in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eace4d51e1342c595aecea0dc97067d : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a48f2516ec143dfbdf4bab2dc52afd4 : FAILURE in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a40f96254eab40cf9e7c06158fac8d38 : FAILURE in 12m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1a0e8612dab40c8ab3c5fd44ced8ccb : FAILURE in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac737b8ed9b14fd991a0f4475636a67d : SUCCESS in 10m 13s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/fb4f94680ec6491bbb831df64cb2ef14 : FAILURE in 1h 05m 15s\n- heat-functional https://zuul.opendev.org/t/openstack/build/7e826350cdd142409733705413528378 : FAILURE in 49m 58s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/91919d52f3bb4743ad529f04d27c2e38 : FAILURE in 1h 01m 25s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa658bdd6496496491e6a98cd877e1f8 : FAILURE in 1h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0fd7d455355a4ed6c5ecf268574afba2f24586c2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-22 13:55:47.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ba53f0b5407175ae605c1f2794d3a4d4e979b7a9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 13:35:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e4f40aa6933ca964c3f8a784613bd16b5def6ea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-22 16:48:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f2f88f05234bebf0fb3ab3bd3c79a0eb60244f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-22 16:48:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"11cc25ba7274b6168621ed1c2da6939597578b9b","revisions":{"a500fae2881bcd2996ac51a22f94fd0bc33ba3b3":{"kind":"REWORK","_number":1,"created":"2023-06-21 01:45:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/886549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/49/886549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/49/886549/1"}}},"commit":{"parents":[{"commit":"04c8faba495b3d0d8ad0c8a1fa174b0ef1357b8d","subject":"db: Removing aliasing of context.session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/04c8faba495b3d0d8ad0c8a1fa174b0ef1357b8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-21 01:43:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-21 01:45:26.000000000","tz":540},"subject":"Stop enabling autocommit","message":"Stop enabling autocommit\n\nWe already fixes the logic dependent on autocommit so we don\u0027t have to\nenable it now.\n\nChange-Id: Ib8a6a1087c2a33155c7471e641ede643cd250852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a500fae2881bcd2996ac51a22f94fd0bc33ba3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a500fae2881bcd2996ac51a22f94fd0bc33ba3b3"}]},"branch":"refs/heads/master"},"1206ecb13608f56ccd20e4e9c1b5bab4a1cb9711":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-06-21 01:46:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/886549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/49/886549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/49/886549/2"}}},"commit":{"parents":[{"commit":"04c8faba495b3d0d8ad0c8a1fa174b0ef1357b8d","subject":"db: Removing aliasing of context.session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/04c8faba495b3d0d8ad0c8a1fa174b0ef1357b8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-21 01:43:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-21 01:45:47.000000000","tz":540},"subject":"db: Disable autocommit","message":"db: Disable autocommit\n\nWe already fixes the logic dependent on autocommit so we don\u0027t have to\nenable it now.\n\nChange-Id: Ib8a6a1087c2a33155c7471e641ede643cd250852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1206ecb13608f56ccd20e4e9c1b5bab4a1cb9711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1206ecb13608f56ccd20e4e9c1b5bab4a1cb9711"}]},"branch":"refs/heads/master"},"d4c0632ce9c02b1a604d8207981b885122974675":{"kind":"REWORK","_number":3,"created":"2023-06-21 02:10:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/49/886549/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/49/886549/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/49/886549/3"}}},"commit":{"parents":[{"commit":"04c8faba495b3d0d8ad0c8a1fa174b0ef1357b8d","subject":"db: Removing aliasing of context.session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/04c8faba495b3d0d8ad0c8a1fa174b0ef1357b8d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-21 01:43:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-21 02:07:47.000000000","tz":540},"subject":"db: Disable autocommit","message":"db: Disable autocommit\n\nWe already fixes the logic dependent on autocommit so we don\u0027t have to\nenable it now.\n\nChange-Id: Ib8a6a1087c2a33155c7471e641ede643cd250852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d4c0632ce9c02b1a604d8207981b885122974675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/d4c0632ce9c02b1a604d8207981b885122974675"}]},"branch":"refs/heads/master"},"11cc25ba7274b6168621ed1c2da6939597578b9b":{"kind":"REWORK","_number":4,"created":"2023-06-22 06:06:46.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/49/886549/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/49/886549/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/49/886549/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/49/886549/4"}}},"commit":{"parents":[{"commit":"2b42bf9788e6057747afd6fa12f045930b545025","subject":"Merge \"db: Removing aliasing of context.session\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2b42bf9788e6057747afd6fa12f045930b545025"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2023-06-22 05:15:04.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2023-06-22 06:06:22.000000000","tz":330},"subject":"db: Disable autocommit","message":"db: Disable autocommit\n\nWe already fixes the logic dependent on autocommit so we don\u0027t have to\nenable it now.\n\nChange-Id: Ib8a6a1087c2a33155c7471e641ede643cd250852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/11cc25ba7274b6168621ed1c2da6939597578b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/11cc25ba7274b6168621ed1c2da6939597578b9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
