)]}'
{"id":"openstack%2Fheat~893673","triplet_id":"openstack%2Fheat~master~I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9","project":"openstack/heat","branch":"master","topic":"aodh-prometheus","attention_set":{"5241":{"account":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"last_update":"2023-11-10 03:23:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2023-10-13 08:27:16.000000000","reason":"\u003cGERRIT_ACCOUNT_8833\u003e replied on the change","reason_account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2023-11-10 00:24:21.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2023-11-10 00:24:21.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-07 02:09:03.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9","subject":"Add OS::Aodh::PrometheusAlarm resource","status":"MERGED","created":"2023-09-04 15:27:54.000000000","updated":"2023-11-10 03:23:47.000000000","submitted":"2023-11-10 00:24:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"893673-aodh-prometheus","meta_rev_id":"a8384585072826d158a909a2d30c3f6d35420c84","_number":893673,"virtual_id_number":893673,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 00:24:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-10-13 08:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-10-04 17:32:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2023-11-09 22:03:00.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2023-11-09 22:03:00.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 17:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 10:41:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-09-19 12:46:22.000000000","updated_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2023-09-25 08:53:10.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2023-09-25 08:53:10.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2023-10-04 17:32:33.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-11-09 22:03:00.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"}],"messages":[{"id":"34881c1ff309cb6b861fc60c021ec5fff606e902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-09-04 15:27:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b028f4fb04d62fae5271a31f1440fe2e83935954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 17:10:33.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f7da790fec1429182d63b387efd460e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/695ba08b8f1d49588df56b79ab0075fb : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/353e1ce7fb86423cb8fbcc91630c03b6 : FAILURE in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28836e70935e4027a324ba66cc88b93d : SUCCESS in 8m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34aac290d793489e9a3a40f4487015f2 : SUCCESS in 8m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3e4e38f137b4eba8308391826374d2b : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe020a7dfb5049a1873366afe96b67b5 : RETRY_LIMIT in 1m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b0effb64804530ba9be7bff310057c : SUCCESS in 11m 03s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/498df74ecd4e438b8896cfa0546c3f23 : SUCCESS in 1h 17m 09s\n- heat-functional https://zuul.opendev.org/t/openstack/build/cd28e7e55c824273adafb07713e1bff9 : FAILURE in 1h 36m 47s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/abe3673514284722a6c0adde89728bdb : FAILURE in 1h 18m 17s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8f4d4f1d78f4979973363863ae78986 : FAILURE in 1h 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65f844d414102ff50e5b6844c35caa8dfd752952","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 17:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a9eff0f8b4d4693a7571e46cbcd2098\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2d85e53d7f94c49a38397f5af855a6e : SUCCESS in 16m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e8e71b00a444ff283b3b1c73902e667 : SUCCESS in 15m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/963d5178da554a30a173742ef2c7569a : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd638d7150fc4026b78e4cd7ea29d47d : RETRY_LIMIT in 2m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8a2913017e9d8d74441f0ec3f668c3106be98cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-09-05 07:37:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b8358aa310a3ee518fe03f18ff795db9d7d7f0dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 08:01:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d0860bfdff43368a6e05d3154211e3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12bb8c98ef1f4df69c7c315caf6f52a3 : SUCCESS in 22m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd0527040f71492f962c421832d498ba : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21c9724e0ab2440695f2dcd78f89e46f : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bfc50705d58a4fb391186b75230be81d : RETRY_LIMIT in 1m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c9c1721b575ae9884c295f637b9cee0f0585bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 08:54:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b6d8d0e8ae34e77851635f26d3a40bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c96efada9f6c4f2494cda0afaa70178b : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39f7834ff194e2480c10b8b56502cd0 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/738ab0b7ca7646138f374076e4f8ee21 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/713b50d088914019a2e66592cb76501f : SUCCESS in 7m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48fb46f91fe14098b10296e41281ffbf : SUCCESS in 8m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03c4b98ccb3041cbbd36e5905316f1bd : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8d89014c4840c4973b6e50438e5f7b : SUCCESS in 12m 34s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/247dd3759f994622ace4de989420c6e5 : SUCCESS in 1h 16m 17s\n- heat-functional https://zuul.opendev.org/t/openstack/build/1bfa8bd368c148cf95e400e8046d13c3 : SUCCESS in 1h 16m 34s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9c2f75e1ac8e4f8b9302396624a2dce4 : SUCCESS in 1h 01m 31s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/164df64f24b649f38573dd77c2048303 : SUCCESS in 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ab74b3ad1f4d1782ef5c39d0d3523736b50fe59","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-06 10:41:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"947f1a3266388584e6f1fba0fa3220a3ec6d810a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-06 10:43:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d472111f0ddf0376825bdf2623739bb9a2c75395","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-09-06 11:51:46.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d4a1dfd319837376accb00798386f2dfe649396c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-09-08 08:33:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a144417211c54e69ddebf811a2de3777b8425114","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 08:48:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54bfbcea64b5494ab5a96a12ccb3daaa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ba8fea0d16745c4b75d68660c52eba4 : SUCCESS in 13m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a5fb40a10e5495880a4714aaaa1a267 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d1473bf854cc42b8a26870b18dbc9557 : SUCCESS in 13m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8338422405dc44ef9a74e7a496149517 : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d1e4689043cbfe135ee0c615ec016dbe68c721ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:26:11.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/891291 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"d6ed14ba18c142ef3adc4150ea7cb9abf07e4d96","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-12 12:46:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b1be9585673e6bb54c2733fd776be3cd524311a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-09-22 13:50:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e418cae9d37a8b64ec87a380f6a80530cafaac16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 15:37:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb42df595f348f29c822f07178f7dd4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d659412b4ba04fa798a7095f49f50129 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5017d30975314cb6bac401143387b9f7 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65d89a2585194030ac4e1691a75c5e20 : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcbc1b5ef9044ae0a2a1c44cf06022b8 : SUCCESS in 9m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9a7b263e88f49ef9125ce91c36743f5 : SUCCESS in 9m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62c2d3d335e344bc89bd360fe441eae4 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e37fd5d7534f4d84f19a57ec45fcb7 : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fc2fd7726b446b682d013bd01e7d79a : SUCCESS in 8m 40s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/39f34aa19b904eadb648f685e7ebda33 : SUCCESS in 1h 41m 26s\n- heat-functional https://zuul.opendev.org/t/openstack/build/73a2e24b779b4503ad6347cf3b41ccd5 : SUCCESS in 1h 14m 29s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/312f074a550043bd87af1b3f02b19d44 : SUCCESS in 1h 05m 18s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb33fcbdedab48afba5739b97bab26ed : SUCCESS in 1h 06m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ec6efc74d0784f131f3e9f908ca9cf0e720bf97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:29:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba28309f69148e6aa5759287fc7083d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b03346afafb4dc087701c05f02e55d7 : SUCCESS in 14m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1150ebc871754ecb85c2cd87ff4c3aba : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01767ba87cf349f69cae5d88c32bfa38 : SUCCESS in 15m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b983a01904a5433788092149fd2e8050 : SUCCESS in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"241c2acc169605d83e9795f727a4ca1e882242c1","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-25 12:43:49.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f5453f2ce0650f027d745174e61216e9a96d7a5b","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-10-02 10:50:49.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1b2a0d7e6b4ac7bb049acd9ff75ee2d51361b5d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-10-02 10:52:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"678c805f049163c0cac358560cac158ea3695c65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:30:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84843ca5155e4de594ee636b662ee003\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd1b0d76ebfb4f7e9068d9704eeb1376 : SUCCESS in 16m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/292d14f24d3547afa5f617d1cc983131 : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26b4b1c0aa6f43a5add40c031622923b : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/96a8c1530a6845c6809940a93de39cbf : SUCCESS in 14m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9d64d3df153d98aaa1783f782d880709fde70536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 12:39:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0b1fe924e1e4a218d0452fc41407683\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bcb819681c1474aa7beba27b2374d2f : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b1fd395e25449e8af28f345583cd82 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/876f1a7edce64537bddc37252cd4a676 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d7c1e16de3a40f59b8c01ead27c5f80 : SUCCESS in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc851df748eb44ee885e34e8f5968c18 : SUCCESS in 7m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f561202198f4fec900ae1553eb49a3b : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c6b2762828d4ab1950236d046e1f66b : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4ddaf519e949e6b8b7077ecefe47fe : SUCCESS in 9m 26s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/853f0b91f19a407b9afa1bf3c283fe70 : SUCCESS in 1h 41m 48s\n- heat-functional https://zuul.opendev.org/t/openstack/build/eb15b1a2cdd84c088b50df1b8eaf982d : SUCCESS in 1h 11m 16s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/ffe9399ced0748d9908729b9a65d73aa : SUCCESS in 1h 13m 00s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/11e95eb5c5ee46e5a989bbb81342d9cf : SUCCESS in 1h 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e93390d1285af581fd87e9a46cb1f86e3ce952c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-03 02:52:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1cd8f49f4a0eb7c0ebdba23e3e6148d63a2dfad7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-03 02:54:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4ee067d91e14b07fd0a99ee8d2b7d93f17ee358b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-03 02:59:37.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d6e12a34cb785b049b0a1dd78eb9be12b1becdfd","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-10-03 11:25:17.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6a75014eca88f5cde6651636031723e2af8f46ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-10-03 11:25:27.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a7fffec50a0a1971b9da694d4f313b9b50485be7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 11:48:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a023aa5aa58a4fedaebac89851e25d97\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af44447cdca34f8d98da359d20427098 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77c4829c14324f4dacfc3559711f1a3b : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e683b80d9164a619c685bd9a9a8ee2f : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da739729f347458eaa4c0584c699e26e : SUCCESS in 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b17072f2cb6984f136680a879cd83ef434d5c38b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 13:07:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe0498ac48d403984638bcf58e31302\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b46b69849def42c4b07ffbd56964f31a : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19959a36118426eb235f3c5c2741261 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe14907e40148e69a862d1bf9b59aa7 : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2fbf9359a5d4ae89b49a05bdeb5aa0b : SUCCESS in 8m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1772618dbecd450faa9ccc7734d67fc1 : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89d48fb8ba7645b88722762ef5e3f709 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7333ddc4371c4217bedc836aa5fcbbe4 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b56e797512b4c49ad8b061ac95757be : SUCCESS in 7m 51s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9747121ac6834c539425303976d92ca3 : SUCCESS in 1h 36m 08s\n- heat-functional https://zuul.opendev.org/t/openstack/build/79f5d2423b9b4138892621e6a87a4e3c : SUCCESS in 1h 13m 20s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9f50c3109c264249be98ac4e350496d5 : SUCCESS in 1h 07m 38s\n- heat-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6f591a2dcf44c94930a5de4d5bd499f : SUCCESS in 1h 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"11632acb8d4b8a931b3d9fc77639b1634bbc86b9","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-10-04 06:06:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"acd3ec25147a50b9ebbe5e132150b324673e7422","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-10-04 17:32:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"15b7650e15a3d0e51f81b9bdba43305756329c36","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-07 02:09:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a34111f1499340dae3996bf109e6821b9a84351e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-10-10 05:59:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6d0a972247d77d2849a4209ef8a2857fb2549766","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-10 06:53:53.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b4b1dfe3b6df3baf69dda2fa12aa82b816bac04c","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-10-13 08:27:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"305c7b7fc0e6a4365dce7e0162b4ce741689e5d8","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2023-11-07 18:21:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0d19fb53be222bc8b86128df6e308e6084db87b3","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2023-11-09 22:03:00.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"139b80e60d90c171dbd0b5ff87a583215fafd040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:03:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a922cb713c5b8a3aa44c9f002a8a16136795b660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 00:24:21.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61a01d8c24a948e4a6a49506a6dd7001\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cb544fc8821434492b58c2f12580f1b : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/effea263185843648e0bceed3079d3d6 : SUCCESS in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cf0941803934f4fa982c7c8fdcded72 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf461764d8894edd94d0895230c96f02 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf5f23d824834757a5057d89fdab1b51 : SUCCESS in 8m 51s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4e82c54eb36a4bd1967ca51398df0f43 : SUCCESS in 2h 13m 27s\n- heat-functional https://zuul.opendev.org/t/openstack/build/9bae94b7172749c284af97f917e9e4df : SUCCESS in 1h 31m 56s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/f4d485f0e29f464ca4815552eec5f07a : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":6},{"id":"68008b4fbb4452653e82ce8190caa37861674777","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 00:24:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"9fe67475de59468c3dfd21941534ea74aede0f23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 00:25:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7849754943bd48e4bbc4b244efe448b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9982f3ddef994cfc83ccda69132e4edc : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11e3791f0d664bce8db9c1378fff1c84 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":6},{"id":"a8384585072826d158a909a2d30c3f6d35420c84","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 03:23:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f8a44f490459fb8ad2700bbbb68dcd2d7deeb488","revisions":{"1847f4c82d9805648cbf4705f1eceb2e3dddfbf5":{"kind":"REWORK","_number":1,"created":"2023-09-04 15:27:54.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/73/893673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/73/893673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/73/893673/1"}}},"commit":{"parents":[{"commit":"a4ac653e35f16a5787bc5a2c934736656294b9b6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a4ac653e35f16a5787bc5a2c934736656294b9b6"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-08-29 13:05:29.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-09-04 15:27:49.000000000","tz":120},"subject":"Add OS::Aodh::PrometheusAlarm resource","message":"Add OS::Aodh::PrometheusAlarm resource\n\nThis patch adds support for Aodh alarms of type Prometheus.\n\nChange-Id: I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1847f4c82d9805648cbf4705f1eceb2e3dddfbf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1847f4c82d9805648cbf4705f1eceb2e3dddfbf5"}]},"branch":"refs/heads/master"},"45d68726d28f067094e610044300fe09d4c525e8":{"kind":"REWORK","_number":2,"created":"2023-09-05 07:37:01.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/73/893673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/73/893673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/73/893673/2"}}},"commit":{"parents":[{"commit":"a4ac653e35f16a5787bc5a2c934736656294b9b6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a4ac653e35f16a5787bc5a2c934736656294b9b6"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-08-29 13:05:29.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-09-05 07:35:57.000000000","tz":120},"subject":"Add OS::Aodh::PrometheusAlarm resource","message":"Add OS::Aodh::PrometheusAlarm resource\n\nThis patch adds support for Aodh alarms of type Prometheus.\n\nChange-Id: I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/45d68726d28f067094e610044300fe09d4c525e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/45d68726d28f067094e610044300fe09d4c525e8"}]},"branch":"refs/heads/master"},"26586470fcb740703d756e142c407269f7b61144":{"kind":"REWORK","_number":3,"created":"2023-09-08 08:33:15.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/73/893673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/73/893673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/73/893673/3"}}},"commit":{"parents":[{"commit":"a4ac653e35f16a5787bc5a2c934736656294b9b6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a4ac653e35f16a5787bc5a2c934736656294b9b6"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-08-29 13:05:29.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-09-08 08:31:10.000000000","tz":120},"subject":"Add OS::Aodh::PrometheusAlarm resource","message":"Add OS::Aodh::PrometheusAlarm resource\n\nThis patch adds support for Aodh alarms of type Prometheus.\n\nDepends-On: Ie7477e0b11479449f7f13c7a784737ac70059964\nChange-Id: I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/26586470fcb740703d756e142c407269f7b61144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/26586470fcb740703d756e142c407269f7b61144"}]},"branch":"refs/heads/master"},"c7f924dbec296942a926635133ebea2c70db32ba":{"kind":"REWORK","_number":4,"created":"2023-09-22 13:50:25.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/73/893673/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/73/893673/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/73/893673/4"}}},"commit":{"parents":[{"commit":"a4ac653e35f16a5787bc5a2c934736656294b9b6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a4ac653e35f16a5787bc5a2c934736656294b9b6"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-08-29 13:05:29.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-09-22 13:46:23.000000000","tz":120},"subject":"Add OS::Aodh::PrometheusAlarm resource","message":"Add OS::Aodh::PrometheusAlarm resource\n\nThis patch adds support for Aodh alarms of type Prometheus.\n\nDepends-On: Ie7477e0b11479449f7f13c7a784737ac70059964\nChange-Id: I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c7f924dbec296942a926635133ebea2c70db32ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/c7f924dbec296942a926635133ebea2c70db32ba"}]},"branch":"refs/heads/master"},"42b200ece80a95bfdf96957a95cfb7b85105a40c":{"kind":"REWORK","_number":5,"created":"2023-10-02 10:52:14.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/73/893673/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/73/893673/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/73/893673/5"}}},"commit":{"parents":[{"commit":"a4ac653e35f16a5787bc5a2c934736656294b9b6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a4ac653e35f16a5787bc5a2c934736656294b9b6"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-08-29 13:05:29.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-10-02 10:52:01.000000000","tz":120},"subject":"Add OS::Aodh::PrometheusAlarm resource","message":"Add OS::Aodh::PrometheusAlarm resource\n\nThis patch adds support for Aodh alarms of type Prometheus.\n\nDepends-On: Ie7477e0b11479449f7f13c7a784737ac70059964\nChange-Id: I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/42b200ece80a95bfdf96957a95cfb7b85105a40c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/42b200ece80a95bfdf96957a95cfb7b85105a40c"}]},"branch":"refs/heads/master"},"f8a44f490459fb8ad2700bbbb68dcd2d7deeb488":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-10-03 11:25:27.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/73/893673/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/73/893673/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/73/893673/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/73/893673/6"}}},"commit":{"parents":[{"commit":"a4ac653e35f16a5787bc5a2c934736656294b9b6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a4ac653e35f16a5787bc5a2c934736656294b9b6"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2023-08-29 13:05:29.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2023-10-03 11:25:27.000000000","tz":0},"subject":"Add OS::Aodh::PrometheusAlarm resource","message":"Add OS::Aodh::PrometheusAlarm resource\n\nThis patch adds support for Aodh alarms of type Prometheus.\n\nDepends-On: I72e124cca4398b78f7ed12e1db3f66bdbfcb196e\nChange-Id: I5bb7c4d9086715fc22c0f7abc36d9bbfc88a60c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f8a44f490459fb8ad2700bbbb68dcd2d7deeb488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f8a44f490459fb8ad2700bbbb68dcd2d7deeb488"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
