)]}'
{"id":"openstack%2Fheat~893945","triplet_id":"openstack%2Fheat~stable%2Fzed~I09805aa4e2d1b43528f70fb29c8981c01deff955","project":"openstack/heat","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2023-09-12 10:17:05.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-12 06:55:21.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I09805aa4e2d1b43528f70fb29c8981c01deff955","subject":"Skip loading auth plugin in case the property is set","status":"MERGED","created":"2023-09-07 03:13:23.000000000","updated":"2023-09-12 10:19:34.000000000","submitted":"2023-09-12 10:17:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893945","cherry_pick_of_change":893714,"cherry_pick_of_patch_set":1,"meta_rev_id":"f5b1a4919f5f9a3ce5aabdd2e68fec838b65f909","_number":893945,"virtual_id_number":893945,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 10:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-09-11 07:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-09-11 06:30:27.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2023-09-11 07:30:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 03:13:23.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-09-07 03:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-07 04:06:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-09-07 04:32:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-11 06:30:27.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"}],"messages":[{"id":"ad9d3988df3b5f853fea8c7b467d8bb8722e1b79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-07 03:13:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"b6ed1c839447ba5fe8833061e6c41a6b64c5894d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 03:23:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69bb63a73bae4e438159a0434503b871\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49a0db7ecb864902b2d6e1c701b9ecf1 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3c6051c89c44c3db10bbbd802d65874 : SUCCESS in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dad1a9e2cf345a4aea08dc91cba6fa741781fb8a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-07 04:06:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a861b8f12cf1897b189c20a886b118f70486100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:32:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f88e1515fa3c47ea823a88497149dc45\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ad2221830e4c37ba5fa08731a2af1f : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9960213c6cd041cfb5570d75ff1c68c8 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d2d961882ac48b4aaebbee1aabc7933 : SUCCESS in 8m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1722912c1bf44596a01dbe20e3633702 : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec7187b8c416410b978f77d6a5d83264 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3abaedb6472f44c6a36d9db27b55a7b8 : SUCCESS in 11m 41s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/4ce81fdb385e4ef8886cc5737a85549e : SUCCESS in 1h 17m 28s\n- heat-functional https://zuul.opendev.org/t/openstack/build/1932d6f4181a40ba93e057337e782f3f : SUCCESS in 1h 11m 06s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/df5e7487bbde440f84db4dfc7f7611d3 : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a0cc9bf22f94bb88205033d622cddd52c5347d1b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-07 04:33:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35e4f6e8c0d2121d1248484ecd78035ee07167d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 04:34:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ba7b2812a3dc7b7c0336f300ccd83cdd32fd66a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 05:49:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3d830969e3849758c8be6dd848848c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdcb641028e04384b22a5c5b2640cba5 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0a54e881e1d4ba69b8b2078f7bf08ea : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dd93079dde643baa5bc0818272f26d4 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/875d0dea939d4284aacc1d96616f0423 : SUCCESS in 12m 09s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/cc5866e41ad34c4c80b1a3448f77cd47 : SUCCESS in 1h 08m 29s\n- heat-functional https://zuul.opendev.org/t/openstack/build/eb0e2223aa834e889ca71caaeb6341bf : FAILURE in 1h 04m 51s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/5fa08f9fbb7f4064907ccb48b6c6e809 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ac0ebf5c0a97a22faade8ee7b2ee2df54b419f3b","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-07 05:53:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d616d7d940d86602774934a28be6b26e5af130c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 06:10:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e2f2d7f1f464281aba97567c6646aba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fca6ed099b2e4629b843f933a024e1c6 : SUCCESS in 15m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69e56f666fc7437ea19fbb47c14df582 : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d6c6d215b66d4ca6a189db217d4b0e6207b2ab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 07:25:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/295f79e398b445229c3b62d3c78b388c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef7d9c1a4f6746a883f3b33b2e831886 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbb1e1a9e38475885c9c9ea788d4287 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/859980de8bd74b6dbdb1c5c81d2218ca : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60654396529e4f239f2993ae1b9578f2 : SUCCESS in 7m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/611384b19bee4014aec060dbfdfb26fa : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01473e2a6c014d7c9adb1f35f4375af8 : SUCCESS in 13m 16s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/8ac5f83ebd4644f89e8c8354c252e940 : SUCCESS in 1h 30m 54s\n- heat-functional https://zuul.opendev.org/t/openstack/build/15188c3831a645fd86194d10ab1de7b4 : SUCCESS in 49m 23s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/c8553d7cb9964944bcb20b4231d57d4b : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c44a2545aa2d851d3349c646d8ee83948f5ebc0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 07:25:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1eff6d120ddd9ab58c81a358fbbc22de9cca04a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 08:35:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/723bc3865dd24edd960fbe09ac295097\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b011dd08c7c469baba42c80c0be644d : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bc62f1ca320416cb86df4e6e4561395 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d83ecf17b3341e58c55b08dbeac9711 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95c68f3883a41c0bbe9893b616598bd : SUCCESS in 13m 20s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/d6046b549ffc48108e4e70d4fe4e4cc2 : FAILURE in 53m 23s\n- heat-functional https://zuul.opendev.org/t/openstack/build/cb6b2594c91f44fb872ac303e5e7eb2b : FAILURE in 57m 20s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/4645241eaa7b481492bbccac6e5d72e0 : FAILURE in 1h 05m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e3d5fc4438c02e4cf9ba107f5b3593540e3deb57","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 00:20:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46cacc1b597f0b9cab8e33e828a4f567f67864a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 00:28:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7649800a98f46cebb029f43ec8007cd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfd2a56fd7f04f03977c8f96ab319109 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46c2ca4f90894e05929c82f5694bcb68 : SUCCESS in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe8db40e283ed78f75ed53dbdbd96fafc899e704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 01:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d7c2b2eaf74f328391199a55454733\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e462b10322844e80987a0d8cde1c5d28 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc0a735a332041f68287b264e3ffb0c0 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdbec83d42da4e9f871a54c5d755e7da : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/283cec75b28e4a88b9cdbfae2360d97a : SUCCESS in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5220a43c43054cff965e6f9926fdf97c : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7bc485c4d943bcb98bde85a9ea9be8 : SUCCESS in 13m 59s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/6f8b900b4341452c9de3604b74bd706d : SUCCESS in 1h 30m 44s\n- heat-functional https://zuul.opendev.org/t/openstack/build/616cdaf021b649499eed850b579f634e : SUCCESS in 52m 10s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7b8b87b68f30468fa08f8234f59e1059 : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f68359133021ff9ab8b5cd1eb662e5b6c55537c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 01:59:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42b1f451dcd0f4d784132ac566e926281b144c91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 03:51:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf024aafc9bb4260b4d4257157ac0f5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b8bb610866b4dd299d4777812cd2d74 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7067a042b9f147c7a9e7e914f5c128ff : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b80d41b51d74ac8b230514e4aa8a952 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a208ba0e3b7f4b7a9c2e3fac7282eefe : SUCCESS in 11m 10s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/39c8f6fe2afa4f2da1d12c51d499faeb : FAILURE in 1h 43m 14s\n- heat-functional https://zuul.opendev.org/t/openstack/build/db9963bbaa5144ec9082c0bfa49c1a41 : SUCCESS in 1h 04m 27s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/7618daa59b1f42a08134120816e4cbe7 : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"92899cf3ce1d645dfad384878635fc8b28da2efe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 10:08:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e442c6eb7e18628f017f0d4e3e7b21c12be58ec8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:27:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f762b2c0475e4009b155cdb8d65184a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/71acd83ce0cc40f487446e9cbe59a6a7 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1daae81651f7402781c542d0ace37c48 : SUCCESS in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dda28149d32f4eac72edf46bf3159844e2a8c28d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:23:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b690499d6a0b4bc19396908926296047\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe2d989b1f3e4000ad8062f8535ed068 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e07d2f0d1e8420ab6d4f0e18e8d7c69 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4774fec33944194b62219401f5865d5 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83e53e096c3e48b98d210a871018d9d9 : SUCCESS in 7m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a305972ed5549e0871f026346c2ab3c : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10738af8ef624931a5e84893927ef586 : SUCCESS in 14m 31s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/58cc6e8293e14fb3ac316ab33aa86a5f : SUCCESS in 1h 13m 16s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d9a3be6f7aaa4b338474441b2a3acedd : FAILURE in 36m 29s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/e99565b5007f481491aa0f0510f588dd : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":1},{"id":"96deddbd4cbd262833f7a639222c2b0fb8d1b0ea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 13:44:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"419a1a6fd4447f2380bf986b5d8d76c8c19ce5cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:02:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07a82c86498a4677b59615ac5ae4c85b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7afa4875cac340c1822b1da8d4956466 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68ca606cc57b494aa25b511aa6ce18df : SUCCESS in 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09be85758691fc607f4ff48646eb7531d5443be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:13:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a8d289af01e4d13ba8adfa3d38aba17\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6adc901e99504377a3163055f9d98917 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb2e1865971f42aa8cf56e99890ded3a : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a4798c872564867b6a3a8b3d9e048a6 : SUCCESS in 8m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/105a8703022e4727af4bd47a62265079 : SUCCESS in 8m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc2be9952c104a76afa87bcf1773f982 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237f909ee1db4e0b90c17892e1db63ee : SUCCESS in 12m 29s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/692e737980c14fbe9bce52144e105a48 : SUCCESS in 1h 24m 56s\n- heat-functional https://zuul.opendev.org/t/openstack/build/8e879a5173794cfbbc499abc43fe1342 : FAILURE in 1h 03m 19s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/d50cb007b06544ddba1fde45099c1420 : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c2d91abfebb651b47c802e30422e8742d7a98dbd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 15:52:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c352ac172d3bd16b1678170162bbccfd0e2b130","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 16:06:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4478815de7c449129cf688a11981d0fa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50da0a6faf764feeac12b043a6de095b : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4330a50baf3b47baac6f7701e3ad7e7f : SUCCESS in 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1577b0b09899da7cfd282fc0f1559ef8679606c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 17:41:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/485c944e1d404d75a16e6aeca487da19\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38caedae0f9b400db2a555b34f3c016f : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e5695fc2634d719b3fde2a5e74af1b : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edd22f85527e4c14a6b40a086496bb75 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc88aa528e2f4454a1f6389219a7d9b2 : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18dc4d0d9ce44a7cb93117f0121d1709 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de4e2e625cf414f870694f34bfb048d : SUCCESS in 10m 06s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/cef822f9c0014fd4836266ef17142ef3 : SUCCESS in 1h 43m 39s\n- heat-functional https://zuul.opendev.org/t/openstack/build/36a3df903eb3479197adb927458ae46b : SUCCESS in 52m 21s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/91b1232877ea40b19f3432bbf3276181 : FAILURE in 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a39873e1bbb7afca6a866e76b36f858d3f6d75d3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 01:53:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00e6ef4ac360a2db9756d1a7361f4d1d7327ecb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 02:10:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28bb218be7e34603850df48d1f3890d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d94f39bd62364dcaac638d3d8992625d : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20294ecaca2844a386ad6759776b4bb5 : SUCCESS in 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96b99d2443497ca6e9c88d5945665d86577a4a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 03:32:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b646b0660941fd8100a21367f054ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/caf817e7073f4a26bb9735d80264e3aa : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718de858ca0e46008f4d3254b7430173 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd695f70545e403e99eb47adb93a8e88 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dee47e8e7e24e16b1e6824b2d75e58d : SUCCESS in 8m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b869e51fbce24714a4b97958b55813a9 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cece5d1cae6e4e64b78aa4bcaee419b7 : SUCCESS in 12m 25s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/dd75eda907fb43f0a970c99c88d003e4 : SUCCESS in 1h 34m 40s\n- heat-functional https://zuul.opendev.org/t/openstack/build/679c9c514f2542268c957cc8248bcb58 : SUCCESS in 1h 06m 41s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/429fa70a87f14b87a5cb42aaa4d7be57 : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ba6c29e24f4def57eb4ec37f24d770f5fd9b1efb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 03:33:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"598de2d1665ddb87eeff1de046854330e9843be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:20:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af832c00085b4064b6b5a72cfc332846\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a731c95d8b4bd4b2db597ee1425793 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2871fa2704b84903b2fd5873bf9ced37 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fff802784b494aa8b93b3902bf041bf8 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3205ab70a0684ef081d394f6ea681c73 : SUCCESS in 13m 32s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/c651ccce9d4d46d79ab0a9fc1eb3204c : SUCCESS in 1h 40m 16s\n- heat-functional https://zuul.opendev.org/t/openstack/build/e568217d2778414a9438d8d4bb985a00 : FAILURE in 46m 12s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/91b8aa0737844de6b2cf31466f792882 : FAILURE in 34m 19s","accounts_in_message":[],"_revision_number":1},{"id":"facaa5a1362100b45cf2ad326281e17919146e1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 05:26:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8375703feed083e518a09234b5857e328d59a896","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 05:44:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac5581f86cdf4f7ea89c29c44caa3308\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12ee1c00a32c482d83141a75f06d1bf8 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8e73a9172474df2a21c5b9ab2ca48f5 : SUCCESS in 15m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c478605f7b73d364feb3c81799a7f1cecd4e11b8","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2023-09-11 06:30:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bcbe06537bac2fc037998a6d5b25d6cf3cd42003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 07:14:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82d502cd7c947a0b95051ae977dfce2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfce65552d704539a4e2d6fdd2404db0 : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e442affbbe6f487cbf209848396edf95 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03f0e23e753446b9bb0fdd9b70a3f558 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49747ba072364ab08d67b8b78fc9f0e4 : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/105b9884c4384fb7a12a8df3f9596dfb : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0dc8cfa0f44196830a00b11279cc6b : SUCCESS in 12m 22s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/9755c2cbe0804a3e8a0a3bd2b700b059 : SUCCESS in 1h 40m 00s\n- heat-functional https://zuul.opendev.org/t/openstack/build/4832a1abce854409a86f71f738bcad74 : SUCCESS in 1h 06m 41s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/8c41020cc15142aca60d5f29d2e4f38c : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":2},{"id":"6d3e4f839a8d180cd4c62a67bdbb9e149f487f5f","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-11 07:30:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c43b7185c904c0026b4130e9d25e898d28be2701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 07:31:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e28b39d386124eb818ebcb1150115f5128613202","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 07:31:48.000000000","message":"Abandoned\n\nthis was merged into https://review.opendev.org/c/openstack/heat/+/894266","accounts_in_message":[],"_revision_number":2},{"id":"e3b7eb1c6edc06e7adf21496a19372fb5a63ebe9","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 07:31:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"f3ca3e25fb33b6381e7e947b949240ae0fcf8cd1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 07:32:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b04ed361e46b8e6f5a916d3f06ee9690088f18ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 08:15:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d7c8ffc22849a8b1113a0b8baeacdb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf6cdf4dd6b94f11ba76b61009492390 : SUCCESS in 15m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ed77f188f1240d09a45e95d71e7abe0 : SUCCESS in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76461a77a2ea1e928f770402c14d98b93d0abc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:15:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/430c6285eb504ee0b44818632b5cb5ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad64b63501794f05a214f999b1cf8c80 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17febb29347841d7a7905753d9c77db4 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c63c35f6fe064a448661ab27b91c9473 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eddef3b8e3b4483898fa63097b973518 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c933f816b5aa4797ad6dbd246b382ab7 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce46aff077343c3ae82127d6dc4f920 : SUCCESS in 12m 03s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/7d9fcfb74da34781b183ba325ac5dfeb : SUCCESS in 1h 37m 34s\n- heat-functional https://zuul.opendev.org/t/openstack/build/dc64f81d0e4c4d14afef7a7cb987fea5 : SUCCESS in 1h 07m 25s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/3e376ef83e30418baf2e882b390e0461 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"71b97d6875b7eceaf3495485149c35c2053a905e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:15:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0739cb6233918422e30e9ebce3fa383127f9b7f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 10:55:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30c18d1d88964c88837eed351a313ec8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1fd31e5f2449279b67a0f8757f437c : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e92ff71492994cefa5d07f938785ce68 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11a8092f8d4a47bb819ba8390fc72246 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9568a6623c4972afc4d3a017c0b5b9 : SUCCESS in 12m 07s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/3ef9a47aa6b0465e9e3463fc2fdc66a4 : SUCCESS in 1h 32m 40s\n- heat-functional https://zuul.opendev.org/t/openstack/build/f565c5d3ff9e4df8a943e438b4a5beb8 : FAILURE in 1h 21m 59s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/fcd628b5121e4c14adc158283a31b314 : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":2},{"id":"171bde680188f148fe21d8aca6389b25c854bcca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 11:01:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6d994ba06e344ff20065d92510c2842a3132d9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 11:26:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cd2a38eb74c4dc8937330c62a9a4220\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/827a548be9d34066bba4748fbe8f5c60 : SUCCESS in 15m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61f1abca05c24af683bdfa24c59efed7 : SUCCESS in 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d58bc975d88f852bf2a2a1a678384b243b233458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 12:23:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19b17c4655d04580a052aed70a92aa7a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/359cc1d9efed4b5190ec197652686a56 : SUCCESS in 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f49ca41259f468895c4c5bb54d943d6 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc69e4a102004f7982ce4af8fc4b6075 : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0b31eca82be4a4fbd1e85a58e2ba102 : SUCCESS in 10m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3f131b4f3114aed913dda5e79595330 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d67cf36dd974ed08b4dc78fd9584cd9 : SUCCESS in 13m 53s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/70446f6cf3254f3a9b20a0cc81904c73 : SUCCESS in 1h 16m 58s\n- heat-functional https://zuul.opendev.org/t/openstack/build/1add54e98d044f309f9fc35c3dd83173 : SUCCESS in 1h 07m 18s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/839525f328bd4c0293c13410f38a4646 : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":2},{"id":"863c9cdfe1437cd544c87129bdbd26589f1fadfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 12:24:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e31be10a50e98acd508f47a9c6509c104dced9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 13:39:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/424874b01709451ea5401f7899bcb246\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/656ad91c8181459692fa883d45f90dcd : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b704d8dbafe74dd3a8cfcbcc63debf91 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16cb8f5c6cbf4c8bb35cc5ecbf976bf0 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f4fc5828b743a78307aefa36ccafa3 : SUCCESS in 10m 40s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/bab921c50eb540e8b91c633a1f92f8bc : SUCCESS in 1h 11m 06s\n- heat-functional https://zuul.opendev.org/t/openstack/build/33abe912cd8f4d0ca8e85d18af2112ee : SUCCESS in 50m 12s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/e4cee5ea63824ec18239a1978917f4d8 : FAILURE in 40m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e85895a85dd925b364aed0c086b52a810b26a398","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 13:46:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e8e594d47a2ff204adbf4a25361d53ed9553ba4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 14:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d5856f3571f4173857eeab5971da36d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8548db78d7d744d29c92bcb7ce01b850 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b6fbeabd0774bc8a33ba0331ae183b6 : SUCCESS in 13m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e92e629c29529d94d727511acb1a63e170032631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 15:12:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/012d43861e1c49f9be523b5024b7937a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e371ec34eef4a94ae9c6a1403365988 : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec047d16ac594c2a870b68f61645989e : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e28e41d8bbd047fd92a3643c905bc9aa : SUCCESS in 10m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1998b973b975421a97fd24e3e10c0f9a : SUCCESS in 9m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/beaef831c87a4438ae04dd70e7d141c6 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b99e79301a84e0cab8e4618526af3d3 : SUCCESS in 15m 12s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/d9af133ba038402c93a9f1a8812f166a : FAILURE in 51m 00s\n- heat-functional https://zuul.opendev.org/t/openstack/build/b7d71af674dd437b931ae03734013c80 : SUCCESS in 1h 09m 36s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/eace0257db6940f5b032bf9b2f8410a2 : SUCCESS in 46m 47s","accounts_in_message":[],"_revision_number":2},{"id":"698561d57294e9f53f0e55afb3c4214e7ebc31c3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-12 06:55:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"497b0152b5d9d0c467c03f7c5d8168539a546e73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 07:13:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1958ab725bb43da99c1b0c799264803\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60ec4214d11e40848c127610aaf1fd23 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/199735396f3f4d4ba3b7adc7b74e4da6 : SUCCESS in 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05f647d5b5642913643a1036fd19b02190f65bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 08:34:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e29be65dea2479cb374237eded6c425\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5addf8e4245f40aeae9f2aba4acdf1b9 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f742813a2a2a40898eee0cd378c8cff3 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7152d0b73a0b4823855eac5c2a07b2eb : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a05d7d860efb4a4da8dc110aa23b2067 : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/169e083c82f24b12ac5563f2e7980bd2 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c7eff7d32074d02b8fd6694f863b772 : SUCCESS in 12m 48s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/94ec2ff7238a4394b1f0f03056fedfa8 : SUCCESS in 1h 32m 21s\n- heat-functional https://zuul.opendev.org/t/openstack/build/6281833e7c004ec48f8f609e030fd6b0 : SUCCESS in 1h 08m 57s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/51d5a76c2e96443086f15426fd4ffbc8 : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c56a6f0473bb8a752b847990d188d7bd7904e14e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 08:35:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2fd266d4fadf4d71c17dcca7ab0420321ee70fe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 10:17:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6416931007f04654a07ff5dfd2b57b03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862d6c0c141b460da3606dca537d938f : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd4a0a7ac1584538940b882e6fb2b2b4 : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7fd6b42a458405a98da675c97055aa4 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5223e13c420f437b91925efe281c2dda : SUCCESS in 10m 22s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/edc4db9d020a4a97a706582565e3d601 : SUCCESS in 1h 36m 12s\n- heat-functional https://zuul.opendev.org/t/openstack/build/4105b0b51234413bbf4e7502b27fb9b6 : SUCCESS in 1h 14m 15s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/897d25dbade74b8083fe104c787d3481 : SUCCESS in 1h 10m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f1e0805c229ee9bd5c8e5e85e879a53668c09772","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 10:17:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f5b1a4919f5f9a3ce5aabdd2e68fec838b65f909","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 10:19:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f90870f4f74534aadbcdcf351fbcaa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e006e304de424393af1f95616d19e5 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2b70a93658ecd2774f22c63a394c5629aefdbe7","revisions":{"bb58b742e5d7e944b1f7cc046f3edcfac4312626":{"kind":"REWORK","_number":1,"created":"2023-09-07 03:13:23.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/45/893945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/45/893945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/893945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/893945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/893945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/45/893945/1"}}},"commit":{"parents":[{"commit":"49b59a8f09cb011a50fa3c11501da07b13f8f3da","subject":"Merge \"Added a unit test for get_environment_hidden_param\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/49b59a8f09cb011a50fa3c11501da07b13f8f3da"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 01:30:57.000000000","tz":540},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2023-09-07 03:13:23.000000000","tz":0},"subject":"Skip loading auth plugin in case the property is set","message":"Skip loading auth plugin in case the property is set\n\nCurrently StoredContext object tries to load auth plugin even when\na specific property(eg. user_domain_id) is already set. However this\nis redundant and we can use the value loaded from stack instead.\n\nBackport note:\nThis change omits the change about GLOBAL_VENV parameter in devstack\nbecause the parameter was added after 2023.1 release.\n\nStory: 2010905\nTask: 48723\nChange-Id: I09805aa4e2d1b43528f70fb29c8981c01deff955\n(cherry picked from commit 7197bd028ba3981d4e6df28885278823015bd5b5)\n(cherry picked from commit afe3e4aa5164064ff328167341d9b0cf15c2587d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/bb58b742e5d7e944b1f7cc046f3edcfac4312626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/bb58b742e5d7e944b1f7cc046f3edcfac4312626"}]},"branch":"refs/heads/stable/zed"},"a2b70a93658ecd2774f22c63a394c5629aefdbe7":{"kind":"REWORK","_number":2,"created":"2023-09-11 05:26:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/45/893945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/45/893945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/893945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/893945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/45/893945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/45/893945/2"}}},"commit":{"parents":[{"commit":"49b59a8f09cb011a50fa3c11501da07b13f8f3da","subject":"Merge \"Added a unit test for get_environment_hidden_param\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/49b59a8f09cb011a50fa3c11501da07b13f8f3da"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 01:30:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 05:26:17.000000000","tz":540},"subject":"Skip loading auth plugin in case the property is set","message":"Skip loading auth plugin in case the property is set\n\nCurrently StoredContext object tries to load auth plugin even when\na specific property(eg. user_domain_id) is already set. However this\nis redundant and we can use the value loaded from stack instead.\n\nBackport note:\nThis change omits the change about GLOBAL_VENV parameter in devstack\nbecause the parameter was added after 2023.1 release.\n\nAlso, this contains change Ie0876080c771fb124d4dd36f803fbfd3b108e240\nbecause Fedora 36 image is being removed from some mirrors.\n\nStory: 2010905\nTask: 48723\nChange-Id: I09805aa4e2d1b43528f70fb29c8981c01deff955\n(cherry picked from commit 7197bd028ba3981d4e6df28885278823015bd5b5)\n(cherry picked from commit afe3e4aa5164064ff328167341d9b0cf15c2587d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a2b70a93658ecd2774f22c63a394c5629aefdbe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/a2b70a93658ecd2774f22c63a394c5629aefdbe7"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
