)]}'
{"id":"openstack%2Fheat~894261","triplet_id":"openstack%2Fheat~stable%2F2023.1~Ie0876080c771fb124d4dd36f803fbfd3b108e240","project":"openstack/heat","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-08 11:12:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0876080c771fb124d4dd36f803fbfd3b108e240","subject":"Bump Fedora image used in CI","status":"MERGED","created":"2023-09-08 03:17:17.000000000","updated":"2023-09-08 11:15:03.000000000","submitted":"2023-09-08 11:12:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894261","cherry_pick_of_change":885725,"cherry_pick_of_patch_set":4,"meta_rev_id":"dc6df4eba28fa2501610f149c39f5f64f09285e0","_number":894261,"virtual_id_number":894261,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 11:12:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-09-08 09:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2023-09-08 09:53:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 03:25:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-08 04:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 09:53:40.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"168089838c7fb51b27bc997bde56573b91ae8379","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 03:17:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"557b795fece51ea23be97c2a9d2f57b24c66803b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 03:25:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cfd4f85558409486743e441557856d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/233fa1191842490ab4fef7fc3573f11d : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e3e4292341645e581c7f7607cfb1d06 : SUCCESS in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4aad498b9f8039068f87a15fb5d77fdf34ee6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 04:44:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e49b311527fc41219c6d36d8920c1df4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aecb191a38c04d62983f78dc3887854e : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62feaa4d99a1476b96ef0053639808e6 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ae202e8fd604248a3e3d13650170cf3 : SUCCESS in 9m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56dc7629a15747b7a54ee8237025b4f6 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfcacfd2e3f4d8ba52cd91c6233d9bd : SUCCESS in 10m 25s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/86820b8bc7b54db7b64e96af31f87d04 : FAILURE in 1h 14m 11s\n- heat-functional https://zuul.opendev.org/t/openstack/build/6b4c8f1bae174621866d9a5fe857e5e4 : SUCCESS in 1h 09m 08s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/9087150ac1b04f3190777f4e3050d765 : FAILURE in 1h 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ddd26790eebdd5c8bbc1e95da8d320562882dd93","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 06:33:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b460603f465b42b387e9d94dd87eedb521c422e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 06:46:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77312d26251348a2a4142c83c74e3eb6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33d0364107234e25a3f169ac60b22a63 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c868ca6b56ac412093bb077b3ce35138 : SUCCESS in 9m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"579edf32af03d3a750078743708102d05ddb39a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 07:50:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe21ad4b2e440acbdc2f51d44e96334\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5edb2eae5c1641bb965f30ae30755ab7 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39df84f3265a4575b57f664f071c061d : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10551f57c76b497f91b50f28d8862ac7 : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f8da13c3b174641931526569c6cd068 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4456f4eedeec49c8a5c3964795f79824 : SUCCESS in 13m 59s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/460a6844d281425c91596a91b9f78ebe : SUCCESS in 1h 15m 31s\n- heat-functional https://zuul.opendev.org/t/openstack/build/3f2fc699f0464d508a22a4975f990a10 : SUCCESS in 1h 02m 20s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/47f294d2fcb7491aa16833811432248a : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f81c01cf303e83f6ea74c38a2175950b902ace44","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-08 09:53:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4eab3d0589cb691081dacb2c86591c9309f588e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:54:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"061d67f9d22c5e1c22a88d8a714203c29dcdb8dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:12:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae5c47092eb42a89cce0f52c2c98c30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c8816e2f3a4bd69eb9ed85a474ba22 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b4abc99906e4d20959167b65b68db85 : SUCCESS in 9m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28429d5951c14c3f8585a742b1cf1f76 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0fba262616642e79a58ee13b8329725 : SUCCESS in 13m 49s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/13bbcf21534247dc8a4e2592de1c5394 : SUCCESS in 1h 17m 02s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d641311dddcc42c484bc91333e53f26e : SUCCESS in 54m 58s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/08dd777453614a9a856c691dad7d19a8 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1422e11ec5eb827e1c71024fe550ec8a20381882","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:12:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dc6df4eba28fa2501610f149c39f5f64f09285e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:15:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbe1c834f78d475ba6576ffe3ef1c29c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ae0064b75648249b8317263ab364ba : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91709d22020c8cdba8fab19a00854e65bde90887","revisions":{"91709d22020c8cdba8fab19a00854e65bde90887":{"kind":"REWORK","_number":1,"created":"2023-09-08 03:17:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/894261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/61/894261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/61/894261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/61/894261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/61/894261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/61/894261/1"}}},"commit":{"parents":[{"commit":"afe3e4aa5164064ff328167341d9b0cf15c2587d","subject":"Skip loading auth plugin in case the property is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/afe3e4aa5164064ff328167341d9b0cf15c2587d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:09:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 03:17:17.000000000","tz":0},"subject":"Bump Fedora image used in CI","message":"Bump Fedora image used in CI\n\nFedora 36 is already EOLed so we should use more recent version.\n\nBecause guest enters to emergency shell when Fedora 38 (latest at\nthe time of writing) is used, we select Fedora 37 for now.\n\nChange-Id: Ie0876080c771fb124d4dd36f803fbfd3b108e240\n(cherry picked from commit 0e515064198f71f8ba388e7fabc93f108faa1891)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/91709d22020c8cdba8fab19a00854e65bde90887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/91709d22020c8cdba8fab19a00854e65bde90887"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
