)]}'
{"id":"openstack%2Fheat~894267","triplet_id":"openstack%2Fheat~stable%2Fyoga~Ie0876080c771fb124d4dd36f803fbfd3b108e240","project":"openstack/heat","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-19 04:29:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie0876080c771fb124d4dd36f803fbfd3b108e240","subject":"Bump Fedora image used in CI","status":"ABANDONED","created":"2023-09-08 11:55:44.000000000","updated":"2023-09-19 04:29:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":894266,"cherry_pick_of_patch_set":1,"meta_rev_id":"9cc36788a48101d22c37fefb0ce5103fd3a080e9","_number":894267,"virtual_id_number":894267,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-08 13:34:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 12:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-08 13:34:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b98f2f4fba6584ad99256fca7f734958dcab8aa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 11:55:44.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"0d2152adf19d0c5ad014f137be1379e995ea3517","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 12:12:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b092c1bee2e7460fa49aa48619d99fbe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba8003ab34d04d199da62e84ba16b23a : SUCCESS in 13m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d31ffa9612144a39af4f96380c628d74 : SUCCESS in 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4f3d34d2ceffde1ad778cc480384cc046a2c93a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:34:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11fce0cf710e418ca83f3ab6411a89e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c1cc476f0b34e8799ffcc5e00dd0851 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96757926c48943e48b0463ad66abe4a5 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f744d20a6e7241b6a5166f95efec0c44 : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea883ec2e5c54eacbf8d9b313e1391eb : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93840d304af0463196abcfd5cfe20f11 : SUCCESS in 9m 43s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/f705c986bb14449fab4a1e6395bfdd55 : SUCCESS in 1h 33m 05s\n- heat-functional https://zuul.opendev.org/t/openstack/build/da8520d9b8514f71bf3d8e184937ff32 : FAILURE in 1h 22m 40s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/414fc43be86a4e3ca032f10ecbf7baeb : FAILURE in 1h 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9cc36788a48101d22c37fefb0ce5103fd3a080e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-19 04:29:33.000000000","message":"Abandoned\n\nthis was merged into https://review.opendev.org/c/openstack/heat/+/894425","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90bd85dcbcc5590b0a6a9d02f9837351a2e8b035","revisions":{"90bd85dcbcc5590b0a6a9d02f9837351a2e8b035":{"kind":"REWORK","_number":1,"created":"2023-09-08 11:55:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/894267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/67/894267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/67/894267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/67/894267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/67/894267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/67/894267/1"}}},"commit":{"parents":[{"commit":"41a6b91362c3cf90bd3692c88b9840d6b1a8d56b","subject":"Added a unit test for get_environment_hidden_param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/41a6b91362c3cf90bd3692c88b9840d6b1a8d56b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:09:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-08 11:55:44.000000000","tz":0},"subject":"Bump Fedora image used in CI","message":"Bump Fedora image used in CI\n\nFedora 36 is already EOLed so we should use more recent version.\n\nBecause guest enters to emergency shell when Fedora 38 (latest at\nthe time of writing) is used, we select Fedora 37 for now.\n\nChange-Id: Ie0876080c771fb124d4dd36f803fbfd3b108e240\n(cherry picked from commit 0e515064198f71f8ba388e7fabc93f108faa1891)\n(cherry picked from commit 91709d22020c8cdba8fab19a00854e65bde90887)\n(cherry picked from commit a89fce30f486fa6e9a55e5873491735725e5174e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/90bd85dcbcc5590b0a6a9d02f9837351a2e8b035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/90bd85dcbcc5590b0a6a9d02f9837351a2e8b035"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
