)]}'
{"id":"openstack%2Fheat~89438","triplet_id":"openstack%2Fheat~stable%2Fhavana~Ib3ec15e6826a40d29a53c7d3afbf12acccd0c245","project":"openstack/heat","branch":"stable/havana","topic":"bump_2013.2.4","hashtags":[],"change_id":"Ib3ec15e6826a40d29a53c7d3afbf12acccd0c245","subject":"Bump stable/havana next version to 2013.2.4","status":"MERGED","created":"2014-04-21 22:35:44.000000000","updated":"2014-05-02 23:42:50.000000000","submitted":"2014-05-02 23:42:50.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb8b37934b346599fa14a4d221f1defab5ea427e","_number":89438,"virtual_id_number":89438,"owner":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-05-02 23:42:50.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-05-02 21:06:31.000000000","_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-05-02 21:06:31.000000000","_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-05-02 23:42:50.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5efc4b1d7e6181498d4b6be34e020df8a6dbda36","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-21 22:35:46.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"91d107f255291258bd2e0ba12ce27bbc1ba846ab","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-21 23:18:05.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-heat-pep8 http://logs.openstack.org/38/89438/1/check/gate-heat-pep8/c31ac38 : FAILURE in 3m 34s\n- gate-heat-docs http://docs-draft.openstack.org/38/89438/1/check/gate-heat-docs/841173c/doc/build/html/ : SUCCESS in 4m 31s\n- gate-heat-python26 http://logs.openstack.org/38/89438/1/check/gate-heat-python26/6500ac6 : SUCCESS in 5m 31s\n- gate-heat-python27 http://logs.openstack.org/38/89438/1/check/gate-heat-python27/e9b18e2 : SUCCESS in 4m 15s\n- check-tempest-dsvm-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-full/a5359a7 : SUCCESS in 36m 47s\n- check-tempest-dsvm-neutron http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-neutron/1b2d22d : SUCCESS in 33m 17s\n- check-devstack-dsvm-cells http://logs.openstack.org/38/89438/1/check/check-devstack-dsvm-cells/87eb1b4 : SUCCESS in 12m 59s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-postgres-full/02b1a3d : SUCCESS in 37m 33s\n","accounts_in_message":[],"_revision_number":1},{"id":"00860f255bf6d2d92f6efaeff9ed4c7a37ad19c5","author":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"date":"2014-04-24 20:35:23.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"45044ebaf4a098cfe680ef21765788d975650722","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-24 20:35:24.000000000","message":"Patch Set 1: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"2239459b55ae8d8b9a1c98be5ef9105d00852db9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-24 21:17:27.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-heat-pep8 http://logs.openstack.org/38/89438/1/check/gate-heat-pep8/d99fe95 : FAILURE in 4m 11s\n- gate-heat-docs http://docs-draft.openstack.org/38/89438/1/check/gate-heat-docs/46d8158/doc/build/html/ : SUCCESS in 4m 20s\n- gate-heat-python26 http://logs.openstack.org/38/89438/1/check/gate-heat-python26/6bad5f4 : SUCCESS in 4m 14s\n- gate-heat-python27 http://logs.openstack.org/38/89438/1/check/gate-heat-python27/a51eb03 : SUCCESS in 5m 01s\n- check-tempest-dsvm-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-full/4184fe4 : SUCCESS in 38m 27s\n- check-tempest-dsvm-neutron http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-neutron/5d82253 : SUCCESS in 38m 26s\n- check-devstack-dsvm-cells http://logs.openstack.org/38/89438/1/check/check-devstack-dsvm-cells/ddae4b4 : SUCCESS in 10m 53s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-postgres-full/5d88951 : SUCCESS in 33m 16s\n","accounts_in_message":[],"_revision_number":1},{"id":"e3c22a94f19d1c146ad57ac54247540b571ad667","author":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"date":"2014-04-29 18:58:33.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"44188cc8b4f479e05362bc1a62a5301b05b60329","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 18:58:34.000000000","message":"Patch Set 1: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"112b5f792794a77dcd86569300e949fa4fe57ddf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-29 19:42:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-heat-pep8 http://logs.openstack.org/38/89438/1/check/gate-heat-pep8/74c1ac8 : FAILURE in 3m 54s\n- gate-heat-docs http://docs-draft.openstack.org/38/89438/1/check/gate-heat-docs/0872c40/doc/build/html/ : SUCCESS in 5m 53s\n- gate-heat-python26 http://logs.openstack.org/38/89438/1/check/gate-heat-python26/c36c7b1 : SUCCESS in 2m 48s\n- gate-heat-python27 http://logs.openstack.org/38/89438/1/check/gate-heat-python27/0659483 : SUCCESS in 4m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-full/c8b8a3d : SUCCESS in 32m 40s\n- check-tempest-dsvm-neutron http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-neutron/acf53ca : SUCCESS in 33m 54s\n- check-devstack-dsvm-cells http://logs.openstack.org/38/89438/1/check/check-devstack-dsvm-cells/c795f33 : SUCCESS in 13m 00s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-postgres-full/7c4efe6 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8d76b50cd25770f62ab22fc16ddf7bed807609f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 19:17:15.000000000","message":"Patch Set 1: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"58d0c66fe7fe3a6b347b589a98e950a66079531a","author":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"date":"2014-05-02 19:17:15.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"d307a028317ae09122007cbc874406d1c56496b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 20:10:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-heat-pep8 http://logs.openstack.org/38/89438/1/check/gate-heat-pep8/8f587e5 : SUCCESS in 3m 08s\n- gate-heat-docs http://docs-draft.openstack.org/38/89438/1/check/gate-heat-docs/0e781fe/doc/build/html/ : SUCCESS in 4m 48s\n- gate-heat-python26 http://logs.openstack.org/38/89438/1/check/gate-heat-python26/70f552d : SUCCESS in 3m 48s\n- gate-heat-python27 http://logs.openstack.org/38/89438/1/check/gate-heat-python27/faf77a7 : SUCCESS in 4m 00s\n- check-tempest-dsvm-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-full/45127eb : SUCCESS in 33m 54s\n- check-tempest-dsvm-neutron http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-neutron/201a0ca : SUCCESS in 33m 52s\n- check-devstack-dsvm-cells http://logs.openstack.org/38/89438/1/check/check-devstack-dsvm-cells/db45fe5 : SUCCESS in 12m 56s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/38/89438/1/check/check-tempest-dsvm-postgres-full/1eb0775 : SUCCESS in 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5928ea43e2ba619c755030b5989c895075fa139b","author":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"date":"2014-05-02 21:06:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e360377b0be2707b34de0e45861612198400d0f7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 21:06:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"093c91676861a8aa74c92ca24f8b3995181f13d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 22:05:05.000000000","message":"Patch Set 1:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"b126ae021ecd2796b66eaac6fdb079ecbe77e6b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 23:42:50.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1},{"id":"fcc50706685444c7aac156686ba6e0b0e52672aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-02 23:42:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded.\n\n- gate-heat-docs http://docs-draft.openstack.org/38/89438/1/gate/gate-heat-docs/b3c5eb2/doc/build/html/ : SUCCESS in 4m 05s\n- gate-heat-pep8 http://logs.openstack.org/38/89438/1/gate/gate-heat-pep8/6098f1b : SUCCESS in 2m 53s\n- gate-heat-python26 http://logs.openstack.org/38/89438/1/gate/gate-heat-python26/ed79fa2 : SUCCESS in 3m 14s\n- gate-heat-python27 http://logs.openstack.org/38/89438/1/gate/gate-heat-python27/946b38d : SUCCESS in 3m 32s\n- gate-tempest-dsvm-full http://logs.openstack.org/38/89438/1/gate/gate-tempest-dsvm-full/e3394f5 : SUCCESS in 42m 10s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/38/89438/1/gate/gate-tempest-dsvm-neutron/fab5d76 : SUCCESS in 31m 23s\n- gate-devstack-dsvm-cells http://logs.openstack.org/38/89438/1/gate/gate-devstack-dsvm-cells/b4a4703 : SUCCESS in 11m 23s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/38/89438/1/gate/gate-tempest-dsvm-postgres-full/e2825c7 : SUCCESS in 33m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b0efa48bec1ff73024467e807ef3c2389ac2909","revisions":{"1b0efa48bec1ff73024467e807ef3c2389ac2909":{"kind":"REWORK","_number":1,"created":"2014-04-21 22:35:44.000000000","uploader":{"_account_id":1420,"name":"Adam Gandelman","email":"adamg@ubuntu.com","username":"gandelman-a"},"ref":"refs/changes/38/89438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/38/89438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/38/89438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/38/89438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/38/89438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/38/89438/1"}}},"commit":{"parents":[{"commit":"527009d0e4a73481459a5246e898390fb432d612","subject":"Don\u0027t delete trust on backup stack delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/527009d0e4a73481459a5246e898390fb432d612"}]}],"author":{"name":"Adam Gandelman","email":"adamg@ubuntu.com","date":"2014-04-21 22:34:20.000000000","tz":-420},"committer":{"name":"Adam Gandelman","email":"adamg@ubuntu.com","date":"2014-04-21 22:34:20.000000000","tz":-420},"subject":"Bump stable/havana next version to 2013.2.4","message":"Bump stable/havana next version to 2013.2.4\n\nChange-Id: Ib3ec15e6826a40d29a53c7d3afbf12acccd0c245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1b0efa48bec1ff73024467e807ef3c2389ac2909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/1b0efa48bec1ff73024467e807ef3c2389ac2909"}]},"branch":"refs/heads/stable/havana"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
