)]}'
{"id":"openstack%2Fheat~894426","triplet_id":"openstack%2Fheat~stable%2Fxena~I09805aa4e2d1b43528f70fb29c8981c01deff955","project":"openstack/heat","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2023-09-14 14:16:28.000000000","reason":"\u003cGERRIT_ACCOUNT_8833\u003e replied on the change","reason_account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-14 15:16:04.000000000","reason":"Change was submitted"},"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2023-09-14 15:16:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09805aa4e2d1b43528f70fb29c8981c01deff955","subject":"Skip loading auth plugin in case the property is set","status":"MERGED","created":"2023-09-11 07:33:27.000000000","updated":"2023-09-14 15:18:24.000000000","submitted":"2023-09-14 15:16:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894426","cherry_pick_of_change":894425,"cherry_pick_of_patch_set":1,"meta_rev_id":"87b1c1578d7cae554ce606181d7742ef3973f60e","_number":894426,"virtual_id_number":894426,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 15:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-09-14 14:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2023-09-14 14:16:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-11 08:54:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-11 09:01:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-13 05:04:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2023-09-13 05:04:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"8af80b968c4fe35d2b54283c37a536c68d207456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-11 07:33:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"4eefd151b4008b14127d4044d8dc20cb776dda3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 08:54:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76af48bf32164551b5e1af73503528aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081751348b2d4c0d94a04fa682b19f83 : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89f25352e136441ebb72ba38600e045e : SUCCESS in 17m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17c6cebe895d73eff76b190e7e594706924093e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:01:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0306691298649558af4fa7ca9de2ed5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06960a027c46469289d92b4d1a3f2ba7 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4fa019378a9482aac1941d8410d06ed : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0263768f71eb4db4aa952b5b95d2e172 : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8b3a91827b24ad3a44bf1547527ad32 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abfe57ccb8e143828db5dfb03411a343 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce19b1620e2c4a95882cd1da88556f69 : SUCCESS in 12m 05s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/5107f18146564549a1f19d52f5b85d32 : FAILURE in 1h 22m 32s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/49ef170be69b4a5b959126efaa95e57d : SUCCESS in 1h 04m 22s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/fc8eecc9e44640b5898cdd71138d8ce6 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b9f549729fb5170a562f79a1adbdef2d1dfe8fce","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-14 14:16:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1226c2cebe8cf08a2f0370ebf319dae7c776c979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 14:17:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74a42ce354194a3232769c14671b1350f0b162fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 15:16:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa752d8ea89f464faf3d6bbb2428c8d0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c4eced6e56488b8251815e8de14b35 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7198dc6b1c1342c089be7e217b8f0cd6 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8fed79de6f64189b4675d0a3eecdeb2 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f650ae097e974945b7fbb055ec465c2a : SUCCESS in 11m 42s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/7f7f025cc8904a23a096e88645c9aa71 : FAILURE in 52m 27s (non-voting)\n- heat-functional https://zuul.opendev.org/t/openstack/build/2d04bf48ddb94e2fa88f924d9f226af9 : SUCCESS in 46m 53s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/49ca4f97ebd1441b904d7568cc15c419 : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d58bb7a7725dc44ed413f448b7b7c63319871a16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 15:16:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87b1c1578d7cae554ce606181d7742ef3973f60e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 15:18:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f331d476694d7db787b9d299d2b1d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae252bea39b54ea79a30b0c6b5afd125 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be70cee559d30e73c36199ee3609bfeff266b473","revisions":{"be70cee559d30e73c36199ee3609bfeff266b473":{"kind":"REWORK","_number":1,"created":"2023-09-11 07:33:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/894426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/26/894426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/26/894426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/26/894426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/26/894426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/26/894426/1"}}},"commit":{"parents":[{"commit":"b9cb496b61b0e804d41e9df1a874becc645b3089","subject":"Merge \"Added a unit test for get_environment_hidden_param\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/b9cb496b61b0e804d41e9df1a874becc645b3089"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 01:30:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-11 07:33:27.000000000","tz":0},"subject":"Skip loading auth plugin in case the property is set","message":"Skip loading auth plugin in case the property is set\n\nCurrently StoredContext object tries to load auth plugin even when\na specific property(eg. user_domain_id) is already set. However this\nis redundant and we can use the value loaded from stack instead.\n\nBackport note:\nThis change omits the change about GLOBAL_VENV parameter in devstack\nbecause the parameter was added after 2023.1 release.\n\nAlso, this contains change Ie0876080c771fb124d4dd36f803fbfd3b108e240\nbecause Fedora 36 image is being removed from some mirrors.\n\nStory: 2010905\nTask: 48723\nChange-Id: I09805aa4e2d1b43528f70fb29c8981c01deff955\n(cherry picked from commit 7197bd028ba3981d4e6df28885278823015bd5b5)\n(cherry picked from commit afe3e4aa5164064ff328167341d9b0cf15c2587d)\n(cherry picked from commit a2b70a93658ecd2774f22c63a394c5629aefdbe7)\n(cherry picked from commit 637d3d50695d7ebc06019d43d0a1f8ca1f3ad275)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/be70cee559d30e73c36199ee3609bfeff266b473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/be70cee559d30e73c36199ee3609bfeff266b473"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
