)]}'
{"id":"openstack%2Fheat~896808","triplet_id":"openstack%2Fheat~stable%2Fwallaby~Icfe18a5d299da4b6b4024374791be53f2b3e41f2","project":"openstack/heat","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2023-10-04 05:28:28.000000000","reason":"Change was submitted"},"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2023-10-04 04:20:53.000000000","reason":"\u003cGERRIT_ACCOUNT_8833\u003e replied on the change","reason_account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}},"hashtags":[],"change_id":"Icfe18a5d299da4b6b4024374791be53f2b3e41f2","subject":"Switch to use \u003cuser/project\u003e_domain_id in context","status":"MERGED","created":"2023-09-28 14:46:53.000000000","updated":"2023-10-04 05:29:41.000000000","submitted":"2023-10-04 05:28:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896808","meta_rev_id":"e3ecfb5679bc8f055258afe200ec0bb3cd0d2024","_number":896808,"virtual_id_number":896808,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 05:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2023-10-04 04:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2023-10-04 01:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2023-10-04 04:20:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 14:46:53.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2023-09-28 15:19:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-28 16:34:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-04 01:39:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-10-04 04:16:33.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"6f929b038b595431fb520ee98774c1490594c1d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-09-28 14:46:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b3ca8c6885b369317e392cbf4169d6dac14c415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:19:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04ebd0063e5542f38492665ef752a645\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2423bb378434209872e6bb21ddfe6b4 : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62fbbfb79be7475ca2d3539c35e3c6b0 : SUCCESS in 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4d4d18c35ab3be154ad91d302778ac20a0e54a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 16:34:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3b8f11082e48d8821f98488ce58dd2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f97e5b5199964307b5ecaa1d0defb003 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22004d08bcb0416da2ee3f6a3bd2271f : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96dfee940b2c40abaac70a739be14671 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3713678a77044c47aa7850816736d9c1 : SUCCESS in 8m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6794f1ab40e4414c965074d68b38fce8 : SUCCESS in 10m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b73c43df6343dc841d984e68cbf319 : SUCCESS in 12m 17s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/3482ecfa5aa241c98283748f7b1cdcbe : SUCCESS in 1h 40m 06s\n- heat-functional https://zuul.opendev.org/t/openstack/build/95b38b63ecb74beebe56c644dd7953c3 : SUCCESS in 1h 13m 02s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/371d4445f5f74a0b85204ac067c95c1a : SUCCESS in 1h 07m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1917038c979aed86e3d4d5ecc77b39c65cebf1c0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-04 01:39:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d990e03841338a0bd8ea5839fd070182b3b882f5","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-10-04 04:20:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"85153db007e1fd3174a820272e967444a602f41f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 04:21:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22e2ffc11d730a92daef454cd53d6fa51e3c1a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 05:28:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00f5eafd81bb4d1ebbfcfbfabf847e6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8007fbd2ada94d50bdb11d468dc57ead : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81fc797ae151418e975799e5caed7481 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf225e7d79de41e6bbb6ddbcf4d3c548 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6dbf49476e24138ad0bf902b4f55252 : SUCCESS in 8m 37s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/1978bcb2ad5e42f69d7de6f5b4c9b46f : SUCCESS in 1h 06m 21s\n- heat-functional https://zuul.opendev.org/t/openstack/build/c0863829a78348c6a9a899d963065f43 : SUCCESS in 47m 29s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/8c18dad9fca94b2b9f2f9409c3e7550c : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ed09c1e4bd4171bfb47a0c5cb56304aa9165c5b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 05:28:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3ecfb5679bc8f055258afe200ec0bb3cd0d2024","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 05:29:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef755a53e677473bbd7fec9245fb2216\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5db5073236a472387203632f3e0ec0c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f809bb7a200da29112559c57e431798f3cacd387","revisions":{"f809bb7a200da29112559c57e431798f3cacd387":{"kind":"REWORK","_number":1,"created":"2023-09-28 14:46:53.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/08/896808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/08/896808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/08/896808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/08/896808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/08/896808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/08/896808/1"}}},"commit":{"parents":[{"commit":"2d5a87d3ae38c43813db8f2779de5e990281640b","subject":"Merge \"Added a unit test for get_environment_hidden_param\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/2d5a87d3ae38c43813db8f2779de5e990281640b"}]}],"author":{"name":"ramishra","email":"ramishra@redhat.com","date":"2021-06-23 05:33:50.000000000","tz":330},"committer":{"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","date":"2023-09-28 14:43:51.000000000","tz":330},"subject":"Switch to use \u003cuser/project\u003e_domain_id in context","message":"Switch to use \u003cuser/project\u003e_domain_id in context\n\nUnreadable logs from oslo.context warnings.\n\n\"Using the \u0027project_domain\u0027 argument is deprecated in version\n\u00272.18\u0027 and will be removed in version \u00273.0\u0027\"\n\nChange-Id: Icfe18a5d299da4b6b4024374791be53f2b3e41f2\n(cherry picked from commit 3cf6c575db764741e729e41d87efdd555dc4400d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f809bb7a200da29112559c57e431798f3cacd387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f809bb7a200da29112559c57e431798f3cacd387"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
