)]}'
{"id":"openstack%2Fheat~896809","triplet_id":"openstack%2Fheat~stable%2Fwallaby~I09805aa4e2d1b43528f70fb29c8981c01deff955","project":"openstack/heat","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2023-11-20 11:39:13.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-09-30 16:24:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I09805aa4e2d1b43528f70fb29c8981c01deff955","subject":"Skip loading auth plugin in case the property is set","status":"ABANDONED","created":"2023-09-28 14:46:53.000000000","updated":"2023-11-20 11:39:13.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"967fdd8ebde8e71feb497188ee1ee6abb5213974","_number":896809,"virtual_id_number":896809,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-28 16:04:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-09-29 05:05:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 14:46:53.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-09-28 15:30:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-28 16:04:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 05:05:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"4fcd870bc762fcc2c12d81578c9b9b33de810f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-09-28 14:46:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a441b81843424f21d2283121e94c75809cc64d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:30:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d961d6a0394228bf1824a996b47bb6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88c4fd14616346288c4228c7b3cb17d0 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b2957c7af504b2dae3f0e2657c4b0fd : SUCCESS in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c38678956907e84bcd7ce32ec7cb7d67fba76836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 16:04:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/404a0f41b98a4133b129000ac7e15999\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70d972a359104e7b90059b1e9fa7c75b : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58aa6d75cf7440b197fdfc3a135fc7d0 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcdb1a51ef82414db1a2f4e3418066a7 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37175673e60148168b94e35379733b90 : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0822bcc17044c0fba18e45345a72249 : SUCCESS in 9m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c48ff928da4a9f9e927ff16f55dd3b : SUCCESS in 13m 06s\n- grenade-heat-multinode https://zuul.opendev.org/t/openstack/build/2f595fea040a4ef1b4baa535395e078e : FAILURE in 47m 20s\n- heat-functional https://zuul.opendev.org/t/openstack/build/d544d5deb90c4b3f8c2ce636dc95c5f7 : SUCCESS in 47m 15s\n- heat-functional-legacy https://zuul.opendev.org/t/openstack/build/c6a13ab716f044f29ff0dbd9befd041e : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":1},{"id":"58ac1220c28d28008fe39377f3067230bd5b18a0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-29 05:05:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b2204ced510c3f5bd776cc408c1a05320e6ee0a","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-09-29 05:41:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e7f3980bf486c0012d0ec965d333fbfba34de6c","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2023-09-29 05:50:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1554a51a50f56e1670e2598a5cd23738c56d4212","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-30 16:24:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"967fdd8ebde8e71feb497188ee1ee6abb5213974","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-20 11:39:13.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/heat/+/897039 was merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcb22cd3bf7df0f6384f33e7c40fdee4eda671a2","revisions":{"fcb22cd3bf7df0f6384f33e7c40fdee4eda671a2":{"kind":"REWORK","_number":1,"created":"2023-09-28 14:46:53.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/09/896809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/heat","ref":"refs/changes/09/896809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/896809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/896809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/heat refs/changes/09/896809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/heat refs/changes/09/896809/1"}}},"commit":{"parents":[{"commit":"f809bb7a200da29112559c57e431798f3cacd387","subject":"Switch to use \u003cuser/project\u003e_domain_id in context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/f809bb7a200da29112559c57e431798f3cacd387"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-09-05 01:30:57.000000000","tz":540},"committer":{"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","date":"2023-09-28 14:44:09.000000000","tz":330},"subject":"Skip loading auth plugin in case the property is set","message":"Skip loading auth plugin in case the property is set\n\nCurrently StoredContext object tries to load auth plugin even when\na specific property(eg. user_domain_id) is already set. However this\nis redundant and we can use the value loaded from stack instead.\n\nBackport note:\nThis change omits the change about GLOBAL_VENV parameter in devstack\nbecause the parameter was added after 2023.1 release.\n\nAlso, this contains change Ie0876080c771fb124d4dd36f803fbfd3b108e240\nbecause Fedora 36 image is being removed from some mirrors.\n\nStory: 2010905\nTask: 48723\nChange-Id: I09805aa4e2d1b43528f70fb29c8981c01deff955\n(cherry picked from commit 7197bd028ba3981d4e6df28885278823015bd5b5)\n(cherry picked from commit afe3e4aa5164064ff328167341d9b0cf15c2587d)\n(cherry picked from commit a2b70a93658ecd2774f22c63a394c5629aefdbe7)\n(cherry picked from commit 637d3d50695d7ebc06019d43d0a1f8ca1f3ad275)\n(cherry picked from commit be70cee559d30e73c36199ee3609bfeff266b473)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/fcb22cd3bf7df0f6384f33e7c40fdee4eda671a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/heat/commit/fcb22cd3bf7df0f6384f33e7c40fdee4eda671a2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
